)]}'
{"id":"openstack%2Freleases~908351","triplet_id":"openstack%2Freleases~master~Ibc68c2e8b8d9f95aac86e9b7564b1bcd7fca499c","project":"openstack/releases","branch":"master","topic":"yoga-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-02-19 15:10:01.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-14 16:17:30.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Ibc68c2e8b8d9f95aac86e9b7564b1bcd7fca499c","subject":"[designate-tempest-plugin] Tag yoga-last","status":"MERGED","created":"2024-02-07 21:13:35.000000000","updated":"2024-02-19 15:10:01.000000000","submitted":"2024-02-19 15:10:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908351-yoga-last","meta_rev_id":"227f901f6c6d60ecbd60e42f4972f636672b1435","_number":908351,"virtual_id_number":908351,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-19 15:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-19 14:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-19 14:45:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-19 07:01:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-02-19 14:55:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-02-19 07:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 21:14:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 21:20:14.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-02-07 21:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-19 14:45:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-02-19 14:55:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"50b44baaa930d8a34a7c96b051fb3a2ba950368f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-07 21:13:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffd3fea6423135a49860cdbdcc4d4a463d16650e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 21:27:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10f2b677d8104777aa57ec326e9fd408\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71172fe47ca64e3090b4e4a758a1d4df : FAILURE in 3m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/519bbddd3f7e4421946a8b7124b618e1 : FAILURE in 3m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a126dd3f51994386981f02c5c87a16d1 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"dc1eaa3024cadb5f2d08c19fba284baedd124e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-08 00:53:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"08c63725fd535d760c22c5494fa7fe428c89ea47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 01:02:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69104e085be44079b174b108a9ffd3fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4d5cf19d7746ee964fe9ddd6e13063 : FAILURE in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5bf5d656e754fb0b7e5a8381af04fc6 : FAILURE in 3m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4e0321d2bb114f8f97d786535138ae75 : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"d32776558dfd97c77edcf509fbbca607e9fb1716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-08 01:05:44.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e860f8b358fecd70919f1aa465153654f232c551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 01:16:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b6352492bc48b0b51c960186832d62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520dc9a9fe024ada9400a1cf4bb8ae69 : SUCCESS in 7m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ded20d611a4745a0b89588c3f069d012 : SUCCESS in 5m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f3816b3d9591476398dd0e4bcf9aeda4 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"73c6c34f84a0aa78ce83dd2caca00eab85c97527","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-14 16:17:30.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82e5059d9b82ac8ffd3fcbd2d6e28fcfce3f4594","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 07:01:01.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":4},{"id":"09d8542e684e08655651466b214233f04e476510","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:01:48.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"32bc53abdd2c8828030932de9e2118a27170ba82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:11:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9be777be2c0f42f1a7bf1dc84c5734da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0822414c3eaf4ccab4869c08e7d6477b : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e61b0fe347c940818b326b2570332a10 : SUCCESS in 4m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e0dadde5509b40f3b3ce91ebe21b1d0f : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"1807cca5cc6ddc35db6b84a93d653f9d543e3865","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-19 14:45:03.000000000","message":"Patch Set 4: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9307c6776e4c26bf3ff74539c82bfb092ec9407b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 14:55:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"97e860926029083702afbea3fa9ef20aba3c3990","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 14:55:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a5e07d57ceaee9c9583695a2ccb6e9a498ad3e03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:10:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e72a51a6d2e40c9acafa05707447f56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c40c674eda5f4a349fc8c5968614e34c : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d13ea7a88074b3db74b5842c8e6cc8c : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4a4615855d2451f95774e25ab03eef8 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":4},{"id":"227f901f6c6d60ecbd60e42f4972f636672b1435","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:10:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"51d4ba597335b0c475f739be650d96f4be6bea0c","revisions":{"27369283065a51b6acebf1db53dbeb44fe56b144":{"kind":"REWORK","_number":1,"created":"2024-02-07 21:13:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/51/908351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/908351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/908351/1"}}},"commit":{"parents":[{"commit":"2d729d808599c0e66b47cdf66ebbcd8d569940f1","subject":"Merge \"[OpenStackSDK] Transition Ussuri to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d729d808599c0e66b47cdf66ebbcd8d569940f1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 21:11:27.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 21:11:27.000000000","tz":-480},"subject":"[designate-tempest-plugin] Tag yoga-last","message":"[designate-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ibc68c2e8b8d9f95aac86e9b7564b1bcd7fca499c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/27369283065a51b6acebf1db53dbeb44fe56b144"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/27369283065a51b6acebf1db53dbeb44fe56b144"}]},"branch":"refs/heads/master"},"376798ae94cdb47f67a5ba6d68501fb69960e527":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-08 00:53:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/51/908351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/908351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/908351/2"}}},"commit":{"parents":[{"commit":"11cbcd02d2923426e9f76caa8456ecf6cd9c364c","subject":"Allow \u003cseries\u003e-last tag to be released for Unmaintained release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11cbcd02d2923426e9f76caa8456ecf6cd9c364c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 21:11:27.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-08 00:53:12.000000000","tz":0},"subject":"[designate-tempest-plugin] Tag yoga-last","message":"[designate-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ibc68c2e8b8d9f95aac86e9b7564b1bcd7fca499c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/376798ae94cdb47f67a5ba6d68501fb69960e527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/376798ae94cdb47f67a5ba6d68501fb69960e527"}]},"branch":"refs/heads/master","description":"Rebase"},"0aee8c7589e084e742b2fded51dab27f957e1d21":{"kind":"REWORK","_number":3,"created":"2024-02-08 01:05:44.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/51/908351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/908351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/908351/3"}}},"commit":{"parents":[{"commit":"11cbcd02d2923426e9f76caa8456ecf6cd9c364c","subject":"Allow \u003cseries\u003e-last tag to be released for Unmaintained release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11cbcd02d2923426e9f76caa8456ecf6cd9c364c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 21:11:27.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-08 01:05:44.000000000","tz":0},"subject":"[designate-tempest-plugin] Tag yoga-last","message":"[designate-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ibc68c2e8b8d9f95aac86e9b7564b1bcd7fca499c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0aee8c7589e084e742b2fded51dab27f957e1d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0aee8c7589e084e742b2fded51dab27f957e1d21"}]},"branch":"refs/heads/master"},"51d4ba597335b0c475f739be650d96f4be6bea0c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-19 07:01:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/51/908351/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/908351/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/908351/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/908351/4"}}},"commit":{"parents":[{"commit":"19fc71f78bfd27fe0e9595a07ce29ee7ad70e451","subject":"Merge \"[tempest] Tag yoga-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19fc71f78bfd27fe0e9595a07ce29ee7ad70e451"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 21:11:27.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-19 07:01:01.000000000","tz":0},"subject":"[designate-tempest-plugin] Tag yoga-last","message":"[designate-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ibc68c2e8b8d9f95aac86e9b7564b1bcd7fca499c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/51d4ba597335b0c475f739be650d96f4be6bea0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/51d4ba597335b0c475f739be650d96f4be6bea0c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
