)]}'
{"id":"openstack%2Freleases~908370","triplet_id":"openstack%2Freleases~master~I3f735c0fca42fb5faf549c22652b793ec962787a","project":"openstack/releases","branch":"master","topic":"yoga-last","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-20 11:21:47.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-03-22 17:14:49.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-03-22 17:22:45.000000000","reason":"Change was submitted"},"13692":{"account":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"last_update":"2024-03-22 17:22:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f735c0fca42fb5faf549c22652b793ec962787a","subject":"[kuryr-tempest-plugin] Tag yoga-last","status":"MERGED","created":"2024-02-07 23:04:21.000000000","updated":"2024-03-22 17:22:45.000000000","submitted":"2024-03-22 17:22:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908370-yoga-last","meta_rev_id":"f6ead20f4f7630fe8a76811c9aecccad1cdb3f6f","_number":908370,"virtual_id_number":908370,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-22 17:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-03-22 17:14:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-03-22 15:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-22 17:14:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-22 17:14:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 23:05:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 23:13:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 23:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"state":"REVIEWER"},{"updated":"2024-02-19 12:08:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-02-19 14:51:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-03-22 15:32:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"6801eb38de9e537ffbc2f55e54a13820a79aad3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-07 23:04:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8567f1fd7840a902b795c48ec1bf713bcca5f5f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 23:13:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2d9e941b77c407eb757f9f2bd94adb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f01d7f9e7b5420abe3e595111574f77 : FAILURE in 2m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b9d4d4eda754707a6e7f553fb9946fd : FAILURE in 3m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/032dc346298e4751b2c5b06eefe8f799 : FAILURE in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a58bbc0112a4fa6e0e9ac9a2f6bb89860dcf6c21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-08 00:53:40.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cd0e9ac07845a02f5fdf5d4f33ad76e6be049de5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 01:01:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/001dc8bcdfbd4643ac59c2032d031e27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48f70787dd249ceb0f21d0113007963 : FAILURE in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2b5797545ff41deb5df26933ef85a6c : FAILURE in 3m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c80b39597035490e8745c0627cf81d71 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9e6ec7231cdf590364d6b6a83d88464283569a9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-08 01:04:53.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"33330442df9b998f9665daa669a065e337a4adec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 01:20:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90099403a981407fbba738e33f697ce4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d3ab0cdc9244d3ab23f6b29b6de656 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e6a38f5590a545d290f1eaa6af992c3a : SUCCESS in 5m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dc2a3fe489e845f6b2e5d60a1ace8b12 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"a780e89a61a6cca4b8002afb4ea121501a0b93fb","author":{"_account_id":13692,"name":"Roman Dobosz","email":"gryf73@gmail.com","username":"gryf"},"date":"2024-02-14 10:35:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"12dc6cd75fb66cdb1976d35b2a0c37c79ae3847e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 07:01:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":4},{"id":"5415643ae2243c3841a0f70296f694b83fb5a726","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:02:05.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"8750ee4fb17cbaf076ecba4d9a98b5eabee8f909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:18:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69db6a431a7d4e7b84a62ac998ec6cce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e389ae59d1a9420aba3357d1bcc4e750 : SUCCESS in 6m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42189709c98d4545aa4c80c09f4242a3 : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/63c9ac404d1a44dcad0d8540ce6f3a00 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":4},{"id":"ec808867bfb9249c379bfb3399f24d86f38e09fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 12:08:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c6ebcf9328a1053a97e41b7e699d71ea68bd043","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-19 14:51:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"26b88fbc24503401d8e70009b6cd8ad5c0de3308","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-19 20:23:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"29b799b77d1e6cdcefb6119580cd5dc73f86deca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-19 20:26:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"9b46bc6263781233ae0a652844d013a98dba3863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 20:52:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f46662b03b9740a3b9a42f00f36fc9b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb945cdc42684ddf866f818fbcfd02b8 : SUCCESS in 6m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/508a90507b814a5480438b7376df79b5 : SUCCESS in 5m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7496152c30fb41f39a6223ee7e79ed4f : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c508cede519585e3fcd03998fbe4c9530cb3b93b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-22 15:32:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"afbceda819241c2367fa87c8e0da1c2b7b2ee76d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-22 17:14:49.000000000","message":"Patch Set 5: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f1b50c8115aa6bae7888721f5a6e43d454c3a78f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:15:10.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7bbbc0ccc71d6d124c122c9d1dba6e70994ef22f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:22:44.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da0e3bdc290541eebfd15db1f1c50929\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb8812ac3db24821a06e05e5ed87e109 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08dbd7e8cc4f467b9063446b2e77282c : SUCCESS in 6m 18s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/543ae04154e54843813dfe4fcf9c673d : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":5},{"id":"f6ead20f4f7630fe8a76811c9aecccad1cdb3f6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:22:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"06658f85a3561754ee27496850ead9e277e67813","revisions":{"f0e65420ab3c38fae9a13a5f8e368df2496831c1":{"kind":"REWORK","_number":1,"created":"2024-02-07 23:04:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/70/908370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/908370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/908370/1"}}},"commit":{"parents":[{"commit":"2d729d808599c0e66b47cdf66ebbcd8d569940f1","subject":"Merge \"[OpenStackSDK] Transition Ussuri to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d729d808599c0e66b47cdf66ebbcd8d569940f1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:03:30.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:03:30.000000000","tz":-480},"subject":"[kuryr-tempest-plugin] Tag yoga-last","message":"[kuryr-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I3f735c0fca42fb5faf549c22652b793ec962787a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e65420ab3c38fae9a13a5f8e368df2496831c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0e65420ab3c38fae9a13a5f8e368df2496831c1"}]},"branch":"refs/heads/master"},"bd96304872d2a57b285e1b348763de6a7ef87dc7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-08 00:53:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/70/908370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/908370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/908370/2"}}},"commit":{"parents":[{"commit":"11cbcd02d2923426e9f76caa8456ecf6cd9c364c","subject":"Allow \u003cseries\u003e-last tag to be released for Unmaintained release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11cbcd02d2923426e9f76caa8456ecf6cd9c364c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:03:30.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-08 00:53:40.000000000","tz":0},"subject":"[kuryr-tempest-plugin] Tag yoga-last","message":"[kuryr-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I3f735c0fca42fb5faf549c22652b793ec962787a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd96304872d2a57b285e1b348763de6a7ef87dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd96304872d2a57b285e1b348763de6a7ef87dc7"}]},"branch":"refs/heads/master","description":"Rebase"},"3a3062b592d4d4ddf478eea074e28bdf9090166b":{"kind":"REWORK","_number":3,"created":"2024-02-08 01:04:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/70/908370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/908370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/908370/3"}}},"commit":{"parents":[{"commit":"11cbcd02d2923426e9f76caa8456ecf6cd9c364c","subject":"Allow \u003cseries\u003e-last tag to be released for Unmaintained release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11cbcd02d2923426e9f76caa8456ecf6cd9c364c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:03:30.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-08 01:04:53.000000000","tz":0},"subject":"[kuryr-tempest-plugin] Tag yoga-last","message":"[kuryr-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I3f735c0fca42fb5faf549c22652b793ec962787a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a3062b592d4d4ddf478eea074e28bdf9090166b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3a3062b592d4d4ddf478eea074e28bdf9090166b"}]},"branch":"refs/heads/master"},"7ed8bac17d962b4b2096ce89b8f70c58f67d881c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-19 07:01:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/70/908370/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/908370/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/908370/4"}}},"commit":{"parents":[{"commit":"19fc71f78bfd27fe0e9595a07ce29ee7ad70e451","subject":"Merge \"[tempest] Tag yoga-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19fc71f78bfd27fe0e9595a07ce29ee7ad70e451"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:03:30.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-19 07:01:21.000000000","tz":0},"subject":"[kuryr-tempest-plugin] Tag yoga-last","message":"[kuryr-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I3f735c0fca42fb5faf549c22652b793ec962787a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ed8bac17d962b4b2096ce89b8f70c58f67d881c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ed8bac17d962b4b2096ce89b8f70c58f67d881c"}]},"branch":"refs/heads/master","description":"Rebase"},"06658f85a3561754ee27496850ead9e277e67813":{"kind":"REWORK","_number":5,"created":"2024-02-19 20:26:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/70/908370/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/70/908370/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/70/908370/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/70/908370/5"}}},"commit":{"parents":[{"commit":"d3affd41e623ffe65c4be3bc654e8af460bc3bc4","subject":"Merge \"[barbican] Transition Ussuri to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d3affd41e623ffe65c4be3bc654e8af460bc3bc4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:03:30.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-19 20:26:11.000000000","tz":-480},"subject":"[kuryr-tempest-plugin] Tag yoga-last","message":"[kuryr-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I3f735c0fca42fb5faf549c22652b793ec962787a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/06658f85a3561754ee27496850ead9e277e67813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/06658f85a3561754ee27496850ead9e277e67813"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
