)]}'
{"id":"openstack%2Freleases~908372","triplet_id":"openstack%2Freleases~master~Ie449442192e78f6d6d51a54b0d60ea20987af046","project":"openstack/releases","branch":"master","topic":"yoga-last","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-02-19 15:09:58.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-02-12 16:05:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-02-19 15:09:58.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-02-13 05:41:42.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"Ie449442192e78f6d6d51a54b0d60ea20987af046","subject":"[cinder-tempest-plugin] Tag yoga-last","status":"MERGED","created":"2024-02-07 23:11:12.000000000","updated":"2024-02-19 15:09:58.000000000","submitted":"2024-02-19 15:09:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908372-yoga-last","meta_rev_id":"352a908729064d220bd8fd81e74947c35f64e447","_number":908372,"virtual_id_number":908372,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-19 15:09:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2024-02-19 07:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-02-19 12:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-02-19 07:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":2,"date":"2024-02-19 14:53:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-19 07:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":1,"date":"2024-02-19 14:53:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-02-19 07:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 23:11:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 23:20:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 23:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"REVIEWER"},{"updated":"2024-02-07 23:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-02-09 13:27:55.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-12 15:57:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-02-19 12:10:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f5dce473421807339bea1ff284b20db068254563","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-07 23:11:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8291fcf40af311e220151fb493f53274ec101b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 23:20:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1c0ae8f01164de0ab334167eb83758a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4beed56e0a348c5aa2d34a8031ec27d : FAILURE in 3m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a0924d5ff3fa44d8994f0e5b2094f587 : FAILURE in 3m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5eea1d6427654945b65313732587e9d0 : FAILURE in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a91d622c0c82c6e06c34ff0b25b3c26065fd07dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-08 00:53:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"09898cae4a2a32cc9f8c6666e0c72d12eb90314f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 01:02:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a981d204b5f484bbd5c7bf1772a31cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e568775f6202434b924a0448632d4e89 : FAILURE in 3m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d011dd3cab824c07b270e451bf438dfe : FAILURE in 3m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6abe4c8f712b4aaa94b7ce9b6e7563ca : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"10e401c48889f8a5d605f2949bf5ebb190069484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-08 01:05:23.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cb9ee0864432d0d02701607766e74999bf40267a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 01:15:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05f43450e30d4f1b8de1d3ae3ae3c21d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb4aa0e09c6947ba81f29cdfb0cdecd9 : SUCCESS in 5m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e9f4d353de0430fbffa900614f800c3 : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/aed064314f97446384f25c179ae4ab7d : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"91814af0d49ce2e6d238c30a3dde813b72747ca5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-09 13:27:55.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"019c7d4f4958f7eea58d7e746b62015d53fc6357","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-12 15:57:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b24da0e999e1e1cae03b9dc5d3763c756c95e5c1","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-02-12 16:05:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dd4724959a7facdae1572be7d03855cde6cc0454","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-13 05:41:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1059e00831064103879a72be43df388bd38cde4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 07:02:14.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":4},{"id":"7e35eda916d9ac57fd321ae7d60fa702d7acf506","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:03:00.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"733da0ef61f90b85bac43f6e112a9778cebcfbc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:15:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b120bbacbda4c4f890c562e51c37567\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c826a287618a4af08906e95c17ab4e38 : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc6cce8b80464bb1b3d1effc9d3f3e30 : SUCCESS in 5m 05s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/83b3d59c080a43b2b96a1bcfab6af0ab : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f1271b0402b1cf67c9b4365aaaadda2a56300145","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 12:10:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fce0fb0aca6a186cfa9991d2cdd3c025086ead90","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-19 14:53:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c405de49f037a3a112cce192953318f49da5837e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 14:53:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"440de1c35350c881ac4bebc480e4662394030711","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:09:58.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e42c5e6d44f4685ba41c14fcf07d7b0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abfd407dfdf04697b7b6206545bd0226 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65139280ce1347e99cee516decf71376 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/801c523d3cca496f88a8a580ab210751 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"352a908729064d220bd8fd81e74947c35f64e447","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 15:09:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6b88afa5baa88e12f703a3d9aa3d020a9ffed042","revisions":{"9932979c70098a42962b07230958aed6a16833db":{"kind":"REWORK","_number":1,"created":"2024-02-07 23:11:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/72/908372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/908372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/908372/1"}}},"commit":{"parents":[{"commit":"2d729d808599c0e66b47cdf66ebbcd8d569940f1","subject":"Merge \"[OpenStackSDK] Transition Ussuri to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d729d808599c0e66b47cdf66ebbcd8d569940f1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:10:36.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:10:36.000000000","tz":-480},"subject":"[cinder-tempest-plugin] Tag yoga-last","message":"[cinder-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ie449442192e78f6d6d51a54b0d60ea20987af046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9932979c70098a42962b07230958aed6a16833db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9932979c70098a42962b07230958aed6a16833db"}]},"branch":"refs/heads/master"},"6fd7d1ce92cb9eb73afed026b5332433efea8774":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-08 00:53:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/72/908372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/908372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/908372/2"}}},"commit":{"parents":[{"commit":"11cbcd02d2923426e9f76caa8456ecf6cd9c364c","subject":"Allow \u003cseries\u003e-last tag to be released for Unmaintained release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11cbcd02d2923426e9f76caa8456ecf6cd9c364c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:10:36.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-08 00:53:35.000000000","tz":0},"subject":"[cinder-tempest-plugin] Tag yoga-last","message":"[cinder-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ie449442192e78f6d6d51a54b0d60ea20987af046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6fd7d1ce92cb9eb73afed026b5332433efea8774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6fd7d1ce92cb9eb73afed026b5332433efea8774"}]},"branch":"refs/heads/master","description":"Rebase"},"c5b86b2a01fb8bbfdfb377c98a53d1735a2c84ca":{"kind":"REWORK","_number":3,"created":"2024-02-08 01:05:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/72/908372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/908372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/908372/3"}}},"commit":{"parents":[{"commit":"11cbcd02d2923426e9f76caa8456ecf6cd9c364c","subject":"Allow \u003cseries\u003e-last tag to be released for Unmaintained release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11cbcd02d2923426e9f76caa8456ecf6cd9c364c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:10:36.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-08 01:05:23.000000000","tz":0},"subject":"[cinder-tempest-plugin] Tag yoga-last","message":"[cinder-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ie449442192e78f6d6d51a54b0d60ea20987af046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5b86b2a01fb8bbfdfb377c98a53d1735a2c84ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c5b86b2a01fb8bbfdfb377c98a53d1735a2c84ca"}]},"branch":"refs/heads/master"},"6b88afa5baa88e12f703a3d9aa3d020a9ffed042":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-19 07:02:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/72/908372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/72/908372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/72/908372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/72/908372/4"}}},"commit":{"parents":[{"commit":"19fc71f78bfd27fe0e9595a07ce29ee7ad70e451","subject":"Merge \"[tempest] Tag yoga-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19fc71f78bfd27fe0e9595a07ce29ee7ad70e451"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:10:36.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-19 07:02:14.000000000","tz":0},"subject":"[cinder-tempest-plugin] Tag yoga-last","message":"[cinder-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and tempest\nplugins are branchless which means master version of Tempest and\nits plugins is used to test the supported stable branches.\n\nOnce any stable branch is moved to unmaintained (previously for\nEM) state then, Tempest and its plugins compatible tag[2][3] needs\nto be released so that we can keep testing the unmaintained stable\nbranches with that tag.\n\nTagging \u0027yoga-last\u0027 as well as a new version with latest hash.\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Ie449442192e78f6d6d51a54b0d60ea20987af046\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b88afa5baa88e12f703a3d9aa3d020a9ffed042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b88afa5baa88e12f703a3d9aa3d020a9ffed042"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
