)]}'
{"id":"openstack%2Freleases~908373","triplet_id":"openstack%2Freleases~master~Iad26e77d32aa69db0d1f8071f9d33d751ac8ea9a","project":"openstack/releases","branch":"master","topic":"yoga-last","attention_set":{},"removed_from_attention_set":{"14107":{"account":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"last_update":"2024-03-22 17:52:55.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-03-22 17:52:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iad26e77d32aa69db0d1f8071f9d33d751ac8ea9a","subject":"[solum-tempest-plugin] Tag yoga-last","status":"MERGED","created":"2024-02-07 23:13:32.000000000","updated":"2024-03-22 17:52:55.000000000","submitted":"2024-03-22 17:52:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"908373-yoga-last","meta_rev_id":"eab2fafd7705621fb2afe61f1e17630da6ff20fd","_number":908373,"virtual_id_number":908373,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-22 17:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-03-22 17:29:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-03-22 15:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-22 17:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-22 17:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"}],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-07 23:14:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-02-07 23:20:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2024-02-07 23:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-22 15:32:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-03-22 17:29:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"3d7cb7214cf2ea6aca60ad7e2a0e939775d03f54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-07 23:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"758af555a65afe869fbeb40aadab45b9da42c0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 23:23:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b05a00d05224ffba5228192984b009b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ec0c7923a2479e99e1a5a0d148f3bc : SUCCESS in 6m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/79ea3e75bf554865a23f95c551c45c6e : FAILURE in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a2a577e081264c0e80a15e9d3d34fbcd : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"424467c92fccee68203aee8c1aa8ecf79ef890a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-08 00:54:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fa61035b979558e9fbf93212c8f1da713fd45639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 01:05:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88fd94108ae34f338b8b8434eb4ae765\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ffe145a8504b0d8382582cc406cee7 : SUCCESS in 6m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8add040af0de435dad1418d216d52e42 : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/27e69786a3d540d5afa8dfa16845f585 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"ce60009550c1d1d23bf6720827e3629911704a20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 07:03:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":3},{"id":"9d64e4251b7e7ed512557396b1e77b684a5dfe95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:19:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a449e9036c499eb66d69016de82ca0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f6d7719adc44c33994c555016eacf0b : SUCCESS in 5m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9e01114b14947da932059d9a3aa6fbb : SUCCESS in 4m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b6422af0d8a2403891138dd6c541922a : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"3e6011529c401a2819955ef59781636367b018ff","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-22 15:32:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"645a3e0a6108995795c25e3545b0d311d8f2441a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-22 17:29:47.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"32c7bd22e1f9b2a71c756d7dc53da7f48392f651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:30:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b7e3fda4e4d01e0d78cbc8b0bf7b82d11f1a8b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:52:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4af8a9ce87bf4ca489cea67f51668828\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf78d4f2d48f49f381b7944bc4f1a49e : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658ba2ea88f544a9b8b8ce17eca5c783 : SUCCESS in 7m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/86d762644f7a40eb8c2b9dc8ffe35a8b : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"eab2fafd7705621fb2afe61f1e17630da6ff20fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:52:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c56c75d9e8525b85eda5c74e58ed3d3a933f2ac1","revisions":{"21367c8c8f10e5c958f66101743325a878f7c46f":{"kind":"REWORK","_number":1,"created":"2024-02-07 23:13:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/908373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/908373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/908373/1"}}},"commit":{"parents":[{"commit":"2d729d808599c0e66b47cdf66ebbcd8d569940f1","subject":"Merge \"[OpenStackSDK] Transition Ussuri to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d729d808599c0e66b47cdf66ebbcd8d569940f1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:13:03.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:13:03.000000000","tz":-480},"subject":"[solum-tempest-plugin] Tag yoga-last","message":"[solum-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version\nof Tempest and its plugins is used to test the supported\nstable branches.\n\nOnce any stable branch is moved to unmaintained (previously\nfor EM) state then, Tempest and its plugins compatible tag[2][3]\nneeds to be released so that we can keep testing the unmaintained\nstable branches with that tag.\n\nTagging \u0027yoga-last\u0027 with the already released 2.7.0 tag hash\n(which is latest hash).\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Iad26e77d32aa69db0d1f8071f9d33d751ac8ea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21367c8c8f10e5c958f66101743325a878f7c46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21367c8c8f10e5c958f66101743325a878f7c46f"}]},"branch":"refs/heads/master"},"e81f0b1cfa6e0d6922cd2e3ff3602e1cc9253220":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-08 00:54:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/908373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/908373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/908373/2"}}},"commit":{"parents":[{"commit":"11cbcd02d2923426e9f76caa8456ecf6cd9c364c","subject":"Allow \u003cseries\u003e-last tag to be released for Unmaintained release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/11cbcd02d2923426e9f76caa8456ecf6cd9c364c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:13:03.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-08 00:54:10.000000000","tz":0},"subject":"[solum-tempest-plugin] Tag yoga-last","message":"[solum-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version\nof Tempest and its plugins is used to test the supported\nstable branches.\n\nOnce any stable branch is moved to unmaintained (previously\nfor EM) state then, Tempest and its plugins compatible tag[2][3]\nneeds to be released so that we can keep testing the unmaintained\nstable branches with that tag.\n\nTagging \u0027yoga-last\u0027 with the already released 2.7.0 tag hash\n(which is latest hash).\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Iad26e77d32aa69db0d1f8071f9d33d751ac8ea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e81f0b1cfa6e0d6922cd2e3ff3602e1cc9253220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e81f0b1cfa6e0d6922cd2e3ff3602e1cc9253220"}]},"branch":"refs/heads/master","description":"Rebase"},"c56c75d9e8525b85eda5c74e58ed3d3a933f2ac1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-19 07:03:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/73/908373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/908373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/908373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/908373/3"}}},"commit":{"parents":[{"commit":"19fc71f78bfd27fe0e9595a07ce29ee7ad70e451","subject":"Merge \"[tempest] Tag yoga-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/19fc71f78bfd27fe0e9595a07ce29ee7ad70e451"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-07 23:13:03.000000000","tz":-480},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-02-19 07:03:49.000000000","tz":0},"subject":"[solum-tempest-plugin] Tag yoga-last","message":"[solum-tempest-plugin] Tag yoga-last\n\nYoga branch is in unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version\nof Tempest and its plugins is used to test the supported\nstable branches.\n\nOnce any stable branch is moved to unmaintained (previously\nfor EM) state then, Tempest and its plugins compatible tag[2][3]\nneeds to be released so that we can keep testing the unmaintained\nstable branches with that tag.\n\nTagging \u0027yoga-last\u0027 with the already released 2.7.0 tag hash\n(which is latest hash).\n\n[1] https://review.opendev.org/q/topic:%22yoga-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: Iad26e77d32aa69db0d1f8071f9d33d751ac8ea9a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c56c75d9e8525b85eda5c74e58ed3d3a933f2ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c56c75d9e8525b85eda5c74e58ed3d3a933f2ac1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
