)]}'
{"id":"openstack%2Freleases~912406","triplet_id":"openstack%2Freleases~master~Idb0b393fa78ddf09ae0b88e55ba2e43c9b0117e9","project":"openstack/releases","branch":"master","topic":"caracal-rc1-deadline","attention_set":{},"removed_from_attention_set":{"35600":{"account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"last_update":"2024-03-15 10:50:57.000000000","reason":"\u003cGERRIT_ACCOUNT_35600\u003e replied on the change","reason_account":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-03-15 14:13:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb0b393fa78ddf09ae0b88e55ba2e43c9b0117e9","subject":"Release mistral-dashboard RC1 for 2024.1 Caracal","status":"MERGED","created":"2024-03-11 11:19:20.000000000","updated":"2024-03-15 14:13:03.000000000","submitted":"2024-03-15 14:13:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912406-caracal-rc1-deadline","meta_rev_id":"cb395575d29ef6b1ffc6c10e64c0975effc037ff","_number":912406,"virtual_id_number":912406,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-15 14:13:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"all":[{"value":1,"date":"2024-03-15 10:50:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":2,"date":"2024-03-15 10:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-03-15 13:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-03-15 13:57:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},{"value":1,"date":"2024-03-15 10:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-03-15 10:51:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-11 11:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"state":"REVIEWER"},{"updated":"2024-03-11 11:26:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-11 11:40:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-15 13:57:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"dfe5d7bba08249cef625e179132de27a9ef10682","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-11 11:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2684363657f76657eb45bde769f6e1964f28b85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 11:40:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4b03384d3c45168d4ef814b31180b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae42b42f8f7f4348abfc97c19067bf74 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4342858b3bb4b44ab1c14caa5322c3b : SUCCESS in 4m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c190da234a694f69948410017249739d : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7e4aaf1bbc887af7d97a443c411d1915adb18cfd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-15 10:25:03.000000000","message":"Patch Set 1: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c25b8eccfd73161c16abd00f4040a48a7b324d7d","author":{"_account_id":35600,"name":"Axel Vanzaghi","display_name":"Axel Vanzaghi","email":"avanzaghi.osf@axellink.fr","username":"axel.vanzaghi","status":"OVHcloud"},"date":"2024-03-15 10:50:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8f698f8b2d71bd25143c0d41cea541277f2c1aa","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 10:51:46.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"24550f9a72ad6cd5e856961a75ac11cb1f24a077","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-15 13:57:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"213ff402cb92980d92491b28558b34eb0dd77ca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 13:57:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dae962319a379bf205b8b34a7a5d5c63aa5a8424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:13:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a89599160cb04f2ba7fd7bd2f55a5e56\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107f188e85824ed4accbb6b3a9310ffd : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3aab9080da94bba80608851380020ad : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2f581d1af104508a0e6d5c8fee4f039 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"cb395575d29ef6b1ffc6c10e64c0975effc037ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:13:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72d7973b85200eef61d07e7d74c352ada4a4b1a7","revisions":{"72d7973b85200eef61d07e7d74c352ada4a4b1a7":{"kind":"REWORK","_number":1,"created":"2024-03-11 11:19:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/912406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/912406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/912406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/912406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/912406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/912406/1"}}},"commit":{"parents":[{"commit":"e8230ff402eea5facd9b3835ececf90de5935445","subject":"Merge \"Add Kolla/Kolla-Ansible/Kayobe Caracal cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8230ff402eea5facd9b3835ececf90de5935445"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-03-11 11:19:16.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-03-11 11:19:16.000000000","tz":60},"subject":"Release mistral-dashboard RC1 for 2024.1 Caracal","message":"Release mistral-dashboard RC1 for 2024.1 Caracal\n\nThis proposed a RC1 release for mistral-dashboard to get any merged changes\nout prior to the RC1 deadline for 2024.1 Caracal going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by after Thursday, March 14th.\n\nChange-Id: Idb0b393fa78ddf09ae0b88e55ba2e43c9b0117e9\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72d7973b85200eef61d07e7d74c352ada4a4b1a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72d7973b85200eef61d07e7d74c352ada4a4b1a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
