)]}'
{"id":"openstack%2Freleases~912461","triplet_id":"openstack%2Freleases~master~Ic764ac56e1c0c16a824be84493779c2a893818e9","project":"openstack/releases","branch":"master","topic":"bug/2056656","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-15 14:58:20.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-03-15 14:58:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic764ac56e1c0c16a824be84493779c2a893818e9","subject":"Create new tooz release for 2024.1","status":"MERGED","created":"2024-03-11 14:40:31.000000000","updated":"2024-03-15 15:18:27.000000000","submitted":"2024-03-15 14:58:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912461-bug/2056656","meta_rev_id":"94ad816a1071061935f87dd438118eb471eb3a2e","_number":912461,"virtual_id_number":912461,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-15 14:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-03-12 08:00:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-03-15 14:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-03-13 14:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-03-15 15:18:27.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":1,"date":"2024-03-13 05:44:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":1,"date":"2024-03-13 01:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-03-15 14:43:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-03-12 01:09:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-11 14:41:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-11 15:20:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2024-03-12 08:00:00.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-03-13 01:23:56.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-03-13 05:44:54.000000000","updated_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"reviewer":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"state":"REVIEWER"},{"updated":"2024-03-13 14:06:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-03-15 14:37:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-03-15 15:18:27.000000000","updated_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"reviewer":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"state":"REVIEWER"}],"messages":[{"id":"fc734d4dcde055f15789d6e8df46e912517209d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-11 14:40:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d2c58d808e08653ea20ada2b7485fca3d7bc4af","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 14:41:31.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"4558bee775e87eae878e201e9afeee77c1d4bdfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 14:51:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75c70ce10f1441c5ae9e500ada8a04c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d52577c4734b1086ef010c8d827ebe : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/90a9fed8c2424c1ebbe066bf1f0e89f0 : FAILURE in 5m 12s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2ec309b080d04dc29b303d8d12c3fb34 : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"236d368851f6f940d16d8187fee2879d598c5c35","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-12 01:02:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c8d7d3aebc3c05aed28d877320bf43ea569f3a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-12 01:08:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a3c844b0f2c9ea696f549dffc1c3b4ddb87e1dde","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 01:09:36.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ec1a6a6607099c1a75684dbe66ae0415bc55f3c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 01:15:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12bdf1f31853420b9d11ed5c0f03a3c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6378535627b442bd97b990f52e33419f : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e508c1a5168f43ccace61b6c95f11e29 : SUCCESS in 4m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0bdfc51579d84cae89df11bffcfd74a9 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"358c1e22ef1d2a07ac4e9d6b66eb1f781887f453","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-12 08:00:00.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d26cd36b863b0b3c96204f1ca730f672899af5c7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-12 15:12:52.000000000","message":"Topic set to bug/2056656","accounts_in_message":[],"_revision_number":2},{"id":"139c5c7718ac29889c0961a05f08101c9f5d4cfe","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-03-13 01:23:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"69aeaa85f104785d7790f19cc8a11f756f31a656","author":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"},"date":"2024-03-13 05:44:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"baf5d8765cbf86f150af3e7c9dbc550b6c76ed29","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-13 14:06:56.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f598aa9ae39e78c41d6e3720a1323893d2aa9578","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-03-15 14:37:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad193ee7a7318e39001ae0e4142477c77d559737","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-15 14:43:09.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78601a27162f6cecbe83fb0f2bb83621acb65f3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:43:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"67ad19a24ed54b4e15de1693e6d0087b0e57e745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:58:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa10cf2a37a044ed90da1f1b6e28b48c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80445a6618844ce0b201cc056b3753de : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54347b9c707146da94c63fabaaaa2873 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c858e20e0e4b43e0be3b8d47d190646d : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"0ad600ac97500d19d99431bb3141532d87f88c4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-15 14:58:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"94ad816a1071061935f87dd438118eb471eb3a2e","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2024-03-15 15:18:27.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d061a6102f3a83f860707218c1fdca41b15242f1","revisions":{"b6b34e12c206ddd4767c46a0362dd69a9d426299":{"kind":"REWORK","_number":1,"created":"2024-03-11 14:40:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/912461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/912461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/912461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/912461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/912461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/912461/1"}}},"commit":{"parents":[{"commit":"ba52f6b43388db90449fffb19baecf0f05a3000e","subject":"Merge \"[magnum] Transition python-magnumclient to EOL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba52f6b43388db90449fffb19baecf0f05a3000e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-11 14:38:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-11 14:40:11.000000000","tz":540},"subject":"Create new tooz release for 2024.1","message":"Create new tooz release for 2024.1\n\nThis proposes a new release of tooz. The new release restores\nthe previous behavior of redis driver, which disables authentication\nfor redis sentinel even when it is enabled for redis, which is required\nby some deployment tools like kolla.\n\nRelated-Bug: #2056656\nDepends-on: https://review.opendev.org/c/openstack/tooz/+/912344\nChange-Id: Ic764ac56e1c0c16a824be84493779c2a893818e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6b34e12c206ddd4767c46a0362dd69a9d426299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6b34e12c206ddd4767c46a0362dd69a9d426299"}]},"branch":"refs/heads/master"},"d061a6102f3a83f860707218c1fdca41b15242f1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-12 01:08:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/912461/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/912461/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/912461/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/912461/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/912461/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/912461/2"}}},"commit":{"parents":[{"commit":"e8230ff402eea5facd9b3835ececf90de5935445","subject":"Merge \"Add Kolla/Kolla-Ansible/Kayobe Caracal cycle highlights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e8230ff402eea5facd9b3835ececf90de5935445"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-11 14:38:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-12 01:08:50.000000000","tz":0},"subject":"Create new tooz release for 2024.1","message":"Create new tooz release for 2024.1\n\nThis proposes a new release of tooz. The new release restores\nthe previous behavior of redis driver, which disables authentication\nfor redis sentinel even when it is enabled for redis, which is required\nby some deployment tools like kolla.\n\nRelated-Bug: #2056656\nDepends-on: https://review.opendev.org/c/openstack/tooz/+/912344\nChange-Id: Ic764ac56e1c0c16a824be84493779c2a893818e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d061a6102f3a83f860707218c1fdca41b15242f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d061a6102f3a83f860707218c1fdca41b15242f1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36216,"name":"WangYun","email":"wangyun@yovole.com","username":"wangyun"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
