)]}'
{"id":"openstack%2Freleases~915997","triplet_id":"openstack%2Freleases~master~Ia58499e33e1d3ae06e95cff06b6c8c6d623ff92a","project":"openstack/releases","branch":"master","topic":"telemetry_tempest","attention_set":{},"removed_from_attention_set":{"5202":{"account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"last_update":"2024-04-19 16:58:35.000000000","reason":"\u003cGERRIT_ACCOUNT_5202\u003e replied on the change","reason_account":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2024-04-25 11:04:04.000000000","reason":"Change was submitted"},"35145":{"account":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"last_update":"2024-04-25 11:04:04.000000000","reason":"Change was submitted"},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2024-04-25 11:04:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia58499e33e1d3ae06e95cff06b6c8c6d623ff92a","subject":"Tag a new release for the telemetry-tempest-plugin","status":"MERGED","created":"2024-04-16 15:03:36.000000000","updated":"2024-04-25 11:04:04.000000000","submitted":"2024-04-25 11:04:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915997-telemetry_tempest","meta_rev_id":"ee4ac8522807ca3a82328328a1b2383717a306da","_number":915997,"virtual_id_number":915997,"owner":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 11:04:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2024-04-19 16:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2024-04-25 10:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-04-24 10:45:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2024-04-25 10:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-04-19 16:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-16 15:04:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-16 15:05:22.000000000","updated_by":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2024-04-16 15:05:22.000000000","updated_by":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2024-04-16 15:05:22.000000000","updated_by":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REVIEWER"},{"updated":"2024-04-16 15:21:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 13:01:22.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2024-04-24 10:45:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-04-25 10:40:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"464ca594200b435a2d0709101c22b05b2c7a5604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"date":"2024-04-16 15:03:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d23aeb5444bbe3bf0f771b90f67d93d0dd588d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 15:21:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7a1fd42ebe4580b228458cc881c6b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac0a6eeebe1498e8dc2ae0dc2870dc5 : SUCCESS in 6m 36s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/2dcd8125841a4dd7ada72ebdebef4946 : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2d8d1aa3f2fe464c88e326dde472a356 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"f600a4e8a819e005351d5beb4b9b323361d75fd3","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-04-17 06:39:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b284cf1bde4c6e375943192ca04c100f26b4960b","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-04-18 14:40:59.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65518af7c3e87aec7544fa72987eef2a5826e4f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"date":"2024-04-19 11:33:15.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b4f691a22c37f73cc13121cd80617d0d73320793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"date":"2024-04-19 11:34:06.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2405e02a064f77a20ce8b9fcb81d77c0cf1bdab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"date":"2024-04-19 11:35:06.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"805b95640ae82aaa358250c664cd8ce6af36b60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 11:46:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33993a24b6884d27bdebb7a8b736cb95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0bf5b078c5400dbb0c4184c903c7a0 : FAILURE in 3m 04s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/8f8c0c809a454d1ea1b04a33e7caf1b4 : FAILURE in 3m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5e456dbba43d4894a31bb555fb319e54 : FAILURE in 3m 27s","accounts_in_message":[],"_revision_number":4},{"id":"fdffad8cea3bc3d1007996e46515f1da70b0f34b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-04-19 13:01:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af82aad4d226a6b9b3fcb717ce233681aa52cd67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-04-19 13:04:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e5af45c7a6fbaeca52cf5d85acf8c84056063343","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2024-04-19 13:04:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"91860a6d85913079b2027f844d87a2f6b8eead62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 13:44:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb817b67491a43e58c8c7c33261144c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715f79c28d024efda721f39e209338bd : SUCCESS in 7m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b9b8a1e72a9e43c49af2e7882cc71876 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dd0acbd2481e4d1db7830625cdd1c680 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":5},{"id":"dfae4886f7c2f2e89b74143afc9219ed2f2f3ae8","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2024-04-19 16:58:35.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b1e7be665c0d627a01fe418957ee448b44029393","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:59:24.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"570b36593826b135b78f20e4fb25ecacc1b5ee4a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-24 10:45:40.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4707204bfeef191bb50bb42180db814fb4bda0a2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-04-25 10:40:43.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1ccd62d14becd52bf1d3edde29c367dc0c140729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 10:41:08.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ad4136bf92c63d654eb7861925cbd926223c1d35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 11:04:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fe830c6a4444b2dbe3689438f4f4472\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0721680bfa748e8a3aaefa248f8cc59 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/477e311dc36f40eeafca44d56c80b888 : SUCCESS in 6m 46s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/903336b10d9a412d944e65bec6d7362f : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":5},{"id":"ee4ac8522807ca3a82328328a1b2383717a306da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 11:04:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d033c3333bdf8bf3ddb836a79c9db29b5857a0bb","revisions":{"e539de2547252bbfa189f8869f8a9a9eadef6e33":{"kind":"REWORK","_number":1,"created":"2024-04-16 15:03:36.000000000","uploader":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"ref":"refs/changes/97/915997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/915997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/915997/1"}}},"commit":{"parents":[{"commit":"ca5e182016d4763dabde6995b32af71f90471e5f","subject":"Merge \"New minor release of Cinder in stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca5e182016d4763dabde6995b32af71f90471e5f"}]}],"author":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-16 15:00:01.000000000","tz":120},"committer":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-16 15:00:01.000000000","tz":120},"subject":"tag a new release for telemetry-tempest-plugin","message":"tag a new release for telemetry-tempest-plugin\n\nChange-Id: Ia58499e33e1d3ae06e95cff06b6c8c6d623ff92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e539de2547252bbfa189f8869f8a9a9eadef6e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e539de2547252bbfa189f8869f8a9a9eadef6e33"}]},"branch":"refs/heads/master"},"911312bed5f04366f78dcaf09ebe2b7f8f70852b":{"kind":"NO_CHANGE","_number":2,"created":"2024-04-19 11:33:15.000000000","uploader":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"ref":"refs/changes/97/915997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/915997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/915997/2"}}},"commit":{"parents":[{"commit":"ca5e182016d4763dabde6995b32af71f90471e5f","subject":"Merge \"New minor release of Cinder in stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca5e182016d4763dabde6995b32af71f90471e5f"}]}],"author":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-16 15:00:01.000000000","tz":120},"committer":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-19 11:29:25.000000000","tz":120},"subject":"tag a new release for telemetry-tempest-plugin","message":"tag a new release for telemetry-tempest-plugin\n\nChange-Id: Ia58499e33e1d3ae06e95cff06b6c8c6d623ff92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/911312bed5f04366f78dcaf09ebe2b7f8f70852b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/911312bed5f04366f78dcaf09ebe2b7f8f70852b"}]},"branch":"refs/heads/master"},"c636f8eb35028fcb653433c073f0e7aee59bacb4":{"kind":"NO_CHANGE","_number":3,"created":"2024-04-19 11:34:06.000000000","uploader":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"ref":"refs/changes/97/915997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/915997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/915997/3"}}},"commit":{"parents":[{"commit":"ca5e182016d4763dabde6995b32af71f90471e5f","subject":"Merge \"New minor release of Cinder in stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca5e182016d4763dabde6995b32af71f90471e5f"}]}],"author":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-16 15:00:01.000000000","tz":120},"committer":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-19 11:33:54.000000000","tz":120},"subject":"tag a new release for telemetry-tempest-plugin","message":"tag a new release for telemetry-tempest-plugin\n\nChange-Id: Ia58499e33e1d3ae06e95cff06b6c8c6d623ff92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c636f8eb35028fcb653433c073f0e7aee59bacb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c636f8eb35028fcb653433c073f0e7aee59bacb4"}]},"branch":"refs/heads/master"},"bf28186054000475a9f8287d643898abdab38764":{"kind":"REWORK","_number":4,"created":"2024-04-19 11:35:06.000000000","uploader":{"_account_id":35145,"name":"marihan girgis","email":"mgirgisf@redhat.com","username":"mgirgisf"},"ref":"refs/changes/97/915997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/915997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/915997/4"}}},"commit":{"parents":[{"commit":"ca5e182016d4763dabde6995b32af71f90471e5f","subject":"Merge \"New minor release of Cinder in stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca5e182016d4763dabde6995b32af71f90471e5f"}]}],"author":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-16 15:00:01.000000000","tz":120},"committer":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-19 11:34:57.000000000","tz":120},"subject":"tag a new release for telemetry-tempest-plugin","message":"tag a new release for telemetry-tempest-plugin\n\nChange-Id: Ia58499e33e1d3ae06e95cff06b6c8c6d623ff92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf28186054000475a9f8287d643898abdab38764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf28186054000475a9f8287d643898abdab38764"}]},"branch":"refs/heads/master"},"d033c3333bdf8bf3ddb836a79c9db29b5857a0bb":{"kind":"REWORK","_number":5,"created":"2024-04-19 13:04:15.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/97/915997/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/97/915997/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/97/915997/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/97/915997/5"}}},"commit":{"parents":[{"commit":"ca5e182016d4763dabde6995b32af71f90471e5f","subject":"Merge \"New minor release of Cinder in stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca5e182016d4763dabde6995b32af71f90471e5f"}]}],"author":{"name":"mgirgisf","email":"mgirgisf@redhat.com","date":"2024-04-16 15:00:01.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2024-04-19 13:03:25.000000000","tz":120},"subject":"Tag a new release for the telemetry-tempest-plugin","message":"Tag a new release for the telemetry-tempest-plugin\n\nMost notable update is the inclusion of the prometheus based tests.\n\nChange-Id: Ia58499e33e1d3ae06e95cff06b6c8c6d623ff92a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d033c3333bdf8bf3ddb836a79c9db29b5857a0bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d033c3333bdf8bf3ddb836a79c9db29b5857a0bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
