)]}'
{"id":"openstack%2Freleases~917781","triplet_id":"openstack%2Freleases~master~I90d9ed61018999552abd53513b9d6c52f16c4ee8","project":"openstack/releases","branch":"master","topic":"swift-eol-feature-branches","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-27 16:57:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I90d9ed61018999552abd53513b9d6c52f16c4ee8","subject":"EOL a bunch of Swift feature branches","status":"ABANDONED","created":"2024-04-30 18:16:09.000000000","updated":"2024-06-27 16:57:40.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e5ad85e6709db61c48dd194b244ddbc71f73b9a5","_number":917781,"virtual_id_number":917781,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-30 19:53:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-05-06 05:28:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-04-30 19:41:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 18:16:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 05:28:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"f12a334dfc3a0fdb986216b509917227e44dd316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-30 18:16:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89da0898cd89fdf518e055d4c71b71b0a525a6c1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 18:16:57.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b2104765035653cb6abeb7a1062ddb8894d5e716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 18:32:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c77d4d81067a4edda022d9801feccccf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ab7d603b1a410399cdb2c6aef4232c : SUCCESS in 5m 38s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/37d8532dc4394da8800d458303bf71d1 : FAILURE in 5m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dc44082c55ce478ea96920b0bd07dce8 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5d91b20cb9d1741c64d3ba6d00d6e0e0777e71db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-30 19:40:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"140aa3ead685348af11717012ad430568eb20de6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 19:41:24.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"010a9cb5b0ed3f36874e4a9447b29c20bda37818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 19:53:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f44effca6cc84af1978cd891f3d1329c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc2da73ba2e4555bd76be4d43bbfeb7 : SUCCESS in 7m 11s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fd86d0ee2b2e4d5f910279de450c277f : SUCCESS in 4m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/35a81ebe3c524b6c91e92c269f93f595 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d1626e14c81f7643135f8eff6dec9d1ee60854a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-06 05:28:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5ad85e6709db61c48dd194b244ddbc71f73b9a5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 16:57:40.000000000","message":"Abandoned\n\nFound another/better way","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"249474a35772656fd7ae4517a6ec2acf034394c0","revisions":{"fb1bc765273042285607d966c0d103646454d5a2":{"kind":"REWORK","_number":1,"created":"2024-04-30 18:16:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/81/917781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/917781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/917781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/917781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/917781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/917781/1"}}},"commit":{"parents":[{"commit":"e9bf76579258be229cece5dc5a1793c45486cd66","subject":"Merge \"Final zed release for Horizon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e9bf76579258be229cece5dc5a1793c45486cd66"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-30 18:07:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-30 18:07:25.000000000","tz":-420},"subject":"EOL a bunch of Swift feature branches","message":"EOL a bunch of Swift feature branches\n\nNone of these branches have seen active development for ages; many of\nthe features these branches were used to develop have even merged.\n\nNote that most of them are *not* pointed at the current tip of the\nbranch -- in 2019 a bunch of OpenDev migration patches were pushed;\nthe immediate parent of such patches is used instead. This is an\nattempt to make to have the tag reflect the point in time that Swift\ndevelopers were last concerning themselves with the branches. The\ndate of the chosen commit is used to decide which cycle\u0027s deliverables\nfile would be appropriate.\n\nChange-Id: I90d9ed61018999552abd53513b9d6c52f16c4ee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb1bc765273042285607d966c0d103646454d5a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fb1bc765273042285607d966c0d103646454d5a2"}]},"branch":"refs/heads/master"},"249474a35772656fd7ae4517a6ec2acf034394c0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-30 19:40:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/81/917781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/81/917781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/917781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/917781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/81/917781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/81/917781/2"}}},"commit":{"parents":[{"commit":"4818795c44fd71c68343d495357c82f7d5718b99","subject":"Enable EOL-ing of feature branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4818795c44fd71c68343d495357c82f7d5718b99"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-30 18:07:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-30 19:40:17.000000000","tz":-420},"subject":"EOL a bunch of Swift feature branches","message":"EOL a bunch of Swift feature branches\n\nNone of these branches have seen active development for ages; many of\nthe features these branches were used to develop have even merged.\n\nNote that most of them are *not* pointed at the current tip of the\nbranch -- in 2019 a bunch of OpenDev migration patches were pushed;\nthe immediate parent of such patches is used instead. This is an\nattempt to make to have the tag reflect the point in time that Swift\ndevelopers were last concerning themselves with the branches. The\ndate of the chosen commit is used to decide which cycle\u0027s deliverables\nfile would be appropriate.\n\nChange-Id: I90d9ed61018999552abd53513b9d6c52f16c4ee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/249474a35772656fd7ae4517a6ec2acf034394c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/249474a35772656fd7ae4517a6ec2acf034394c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
