)]}'
{"id":"openstack%2Freleases~920400","triplet_id":"openstack%2Freleases~master~I302050ed62e62b05f09aa8c320dd064edbdb1075","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"8064":{"account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"last_update":"2024-05-29 11:53:01.000000000","reason":"\u003cGERRIT_ACCOUNT_8064\u003e replied on the change","reason_account":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-06-06 08:27:11.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-05-29 12:16:02.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"14394":{"account":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"last_update":"2024-06-06 08:27:11.000000000","reason":"Change was submitted"},"33467":{"account":{"_account_id":33467,"name":"Matt Pryor","email":"matt@stackhpc.com","username":"mkjpryor"},"last_update":"2024-06-06 08:27:11.000000000","reason":"Change was submitted"},"36412":{"account":{"_account_id":36412,"name":"Scott Davidson","email":"sdavidson327@gmail.com","username":"sd109"},"last_update":"2024-06-06 08:27:11.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2024-06-06 08:27:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I302050ed62e62b05f09aa8c320dd064edbdb1075","subject":"magnum-capi-helm: Add 1.0.0 release","status":"MERGED","created":"2024-05-24 15:25:22.000000000","updated":"2024-06-06 08:27:11.000000000","submitted":"2024-06-06 08:27:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920400","meta_rev_id":"e0ed3d87973bcf760136a1cb78024b726cf4a1f0","_number":920400,"virtual_id_number":920400,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-06 08:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"all":[{"value":1,"date":"2024-05-29 11:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":2,"date":"2024-06-06 07:54:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-05-28 15:56:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-05-29 12:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":1,"date":"2024-06-06 07:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-05-28 15:57:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"_account_id":33467,"name":"Matt Pryor","email":"matt@stackhpc.com","username":"mkjpryor"},{"_account_id":36412,"name":"Scott Davidson","email":"sdavidson327@gmail.com","username":"sd109"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":33467,"name":"Matt Pryor","email":"matt@stackhpc.com","username":"mkjpryor"},{"_account_id":36412,"name":"Scott Davidson","email":"sdavidson327@gmail.com","username":"sd109"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-24 15:26:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14394,"name":"Dale Smith","email":"dale@catalystcloud.nz","username":"dalees"},"state":"REVIEWER"},{"updated":"2024-05-24 15:26:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"state":"REVIEWER"},{"updated":"2024-05-24 15:26:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2024-05-24 15:26:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":33467,"name":"Matt Pryor","email":"matt@stackhpc.com","username":"mkjpryor"},"state":"REVIEWER"},{"updated":"2024-05-24 15:26:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":36412,"name":"Scott Davidson","email":"sdavidson327@gmail.com","username":"sd109"},"state":"REVIEWER"},{"updated":"2024-05-24 15:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-24 16:06:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-05-28 10:14:16.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-06-06 07:54:29.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c6ced8318d30144a02897b9e737ecce5f4b2b4a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-24 15:25:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9ff01a0273f5af99989cadc3f0cb287b41f323b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 15:26:25.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2a4323bfe18cff0d09c5de558f7ca05331565314","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-24 15:26:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81052fa216fc4afaaba0519468af436931814515","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-24 15:27:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0699dab8ecf991e01e87ef2542377211c23ddc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 15:52:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af8d6b09d32c4bf2adcfb6da41820995\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b8ad4d252f4383a7b39f8805110ff3 : SUCCESS in 6m 36s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b4f33e3b75e745ea944136ef853e3105 : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1bbc5c2fa23f489fb6b4aaf36c68bbb5 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1035b55aede53441af2cdd54698ca9613473760f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-05-24 16:06:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"643e3dbcb8691389706d7a49b47c838a984495ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-26 05:54:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f0132ea0bd5ec519a19f5cc2e1fc60dedffa1dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 08:35:25.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6edeea27bc2bb95a219c4594baec804ef7e6aa69","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 08:36:19.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"58020af469653cb585906ca59d58f49071149a45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 08:36:45.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":3},{"id":"f1b47fb8512b8c77a2127c75d58d2acdc7d3f3cf","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 08:37:36.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"8725a27f2a0bce18fa5eefbc4231a368b93ed410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 08:52:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c0660737314b4bb40ebf23d3e40924\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5560c920c26943d4a34b16389796a11c : SUCCESS in 7m 03s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6a23d5db39684e8691f5df66a030c98a : SUCCESS in 4m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4698b1a99c4b49adaaa9e5da33c11659 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7fb47c771ace33c78dc164c94bb7e6c7de3cc69b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-05-28 10:14:16.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6fb25bf0f34aad2cf7aef533eb6b42c6522052bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36412,"name":"Scott Davidson","email":"sdavidson327@gmail.com","username":"sd109"},"date":"2024-05-28 15:55:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ad9e7b74597d84e52180f5a91a398eb5d91f0239","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 15:56:40.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d4d37468947ca06e83fb40f2f7a3f6368652a916","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 15:57:29.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"271360c95aa2272d4db9adb61ac7c9338e640e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 16:25:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e5b43f7cbdc44fa8adaebbccb933849\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcfdf55812146b4aa2034cea345648e : SUCCESS in 5m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/25d62b70bfc54633a1a978e99a3097a8 : SUCCESS in 5m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a6a7c5a4938f4a1ea748a719e24a3d24 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"129ed4476881dfa5628628b19f53145550534bf9","author":{"_account_id":8064,"name":"Jake Yip","email":"jake.yip@ardc.edu.au","username":"jake"},"date":"2024-05-29 11:53:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5c2ada7405543632c769fc0f7a0a02e431b03563","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-05-29 12:16:02.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fa0d78d99116283cffb5bca6dad8812b5e87a594","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-06 07:54:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c927e5311425f6082587b4baf5aa60aa29469f05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 07:54:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6bf9fea706d7e19bda4d88dca59950c5b8202456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 08:27:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c2f1504ec6947d6a81956816cc13827\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46cc9a49c96f450c81a6817c4b6efc9a : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587530e7437841f4915946853a6af54c : SUCCESS in 6m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/4f090c2675c240d690332ac028145d46 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e0ed3d87973bcf760136a1cb78024b726cf4a1f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 08:27:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"13cb2ebef324110f4590fc738b877cc8c72ef261","revisions":{"bee15c71437178d8d3b3de0655dc941d2ff0c16e":{"kind":"REWORK","_number":1,"created":"2024-05-24 15:25:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/920400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/920400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/920400/1"}}},"commit":{"parents":[{"commit":"e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8","subject":"Merge \"Release OpenStack-Ansible Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-24 15:23:48.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-24 15:25:20.000000000","tz":0},"subject":"magnum-capi-helm: Add 1.0.0 release","message":"magnum-capi-helm: Add 1.0.0 release\n\nTo distinct from versions/tags released at [1] (latest is 0.13.0)\nwe start with 1.0.0.\n\nChange-Id: I302050ed62e62b05f09aa8c320dd064edbdb1075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bee15c71437178d8d3b3de0655dc941d2ff0c16e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bee15c71437178d8d3b3de0655dc941d2ff0c16e"}]},"branch":"refs/heads/master"},"84b34fa882366215dfa95cb4f98c644e279e0f4c":{"kind":"REWORK","_number":2,"created":"2024-05-28 08:35:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/920400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/920400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/920400/2"}}},"commit":{"parents":[{"commit":"e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8","subject":"Merge \"Release OpenStack-Ansible Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-24 15:23:48.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 08:35:25.000000000","tz":0},"subject":"magnum-capi-helm: Add 1.0.0 release","message":"magnum-capi-helm: Add 1.0.0 release\n\nTo distinct from versions/tags released at [1] (latest is 0.13.0)\nwe start with 1.0.0.\n\nChange-Id: I302050ed62e62b05f09aa8c320dd064edbdb1075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84b34fa882366215dfa95cb4f98c644e279e0f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/84b34fa882366215dfa95cb4f98c644e279e0f4c"}]},"branch":"refs/heads/master"},"8bdeec32948cb2373fed843e9ae7c1f2fb0b5410":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-05-28 08:36:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/920400/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/920400/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/920400/3"}}},"commit":{"parents":[{"commit":"e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8","subject":"Merge \"Release OpenStack-Ansible Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-24 15:23:48.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 08:36:45.000000000","tz":0},"subject":"magnum-capi-helm: Add 1.0.0 release","message":"magnum-capi-helm: Add 1.0.0 release\n\nTo distinct from versions/tags released at [1] (latest is 0.13.0)\nwe start with 1.0.0.\n\n[1]: https://github.com/stackhpc/magnum-capi-helm\n\nChange-Id: I302050ed62e62b05f09aa8c320dd064edbdb1075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8bdeec32948cb2373fed843e9ae7c1f2fb0b5410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8bdeec32948cb2373fed843e9ae7c1f2fb0b5410"}]},"branch":"refs/heads/master","description":"Edit commit message"},"13cb2ebef324110f4590fc738b877cc8c72ef261":{"kind":"REWORK","_number":4,"created":"2024-05-28 15:55:39.000000000","uploader":{"_account_id":36412,"name":"Scott Davidson","email":"sdavidson327@gmail.com","username":"sd109"},"ref":"refs/changes/00/920400/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/00/920400/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/00/920400/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/00/920400/4"}}},"commit":{"parents":[{"commit":"e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8","subject":"Merge \"Release OpenStack-Ansible Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e7de4d2c86708bce9b7aa0c0da3bc828c05a3de8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-24 15:23:48.000000000","tz":0},"committer":{"name":"sd109","email":"scott@stackhpc.com","date":"2024-05-28 15:55:37.000000000","tz":60},"subject":"magnum-capi-helm: Add 1.0.0 release","message":"magnum-capi-helm: Add 1.0.0 release\n\nTo distinct from versions/tags released at [1] (latest is 0.13.0)\nwe start with 1.0.0.\n\n[1]: https://github.com/stackhpc/magnum-capi-helm\n\nChange-Id: I302050ed62e62b05f09aa8c320dd064edbdb1075\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13cb2ebef324110f4590fc738b877cc8c72ef261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13cb2ebef324110f4590fc738b877cc8c72ef261"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
