)]}'
{"id":"openstack%2Freleases~923487","triplet_id":"openstack%2Freleases~master~I5b77a3e6ee9ee0b91c6107e127415e966ebed23a","project":"openstack/releases","branch":"master","topic":"dalmatian-milestone-2","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-07-05 13:35:40.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"13114":{"account":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"last_update":"2024-07-11 04:18:23.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5b77a3e6ee9ee0b91c6107e127415e966ebed23a","subject":"Release monasca for Dalmatian-2 milestone","status":"ABANDONED","created":"2024-07-04 10:15:28.000000000","updated":"2024-07-11 04:18:23.000000000","total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"cdd46daf0280b5d6162289152fae5075a64d08cc","_number":923487,"virtual_id_number":923487,"owner":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-05 10:51:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-07-05 14:11:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 10:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-04 11:02:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 09:55:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-07-05 14:11:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"cbaf04499dc31937aab4273222169f086ec8d508","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-04 10:15:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c857bd9dea25bbc1ec7854c7e962664397982bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 11:02:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/998333cf726a425ebcd835a16c0c71ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e273967f4c544fc89b45c311226aeef : SUCCESS in 6m 52s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1672e9c263c94ba1bb781d4fe3da3bc9 : FAILURE in 10m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cac55c654cac4e499b94bf9ffd944f0f : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a5fb71e26e9af2ddc4b75fecaf08b3590e89f037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-05 07:57:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cbbe2275a380d0e2b430ecf0c54dbb94c21e5c09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:05:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e07af52ead744898bacd69157287635f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/970b7437309e43e89c74daa5942df1f9 : SUCCESS in 6m 43s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/55f5386883d744f3984d82bf3b6c2a07 : FAILURE in 6m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5bb1a12637d94a058c8552240cc251d3 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4b12971f87702d47c2a9db6a7e39721a6619ce29","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-05 09:55:13.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e8c3d47c277b586a1766889ff7fce72654e21f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-05 10:28:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"826c4f3dbb2701e05bbd0bd5d1ac83cff8b6a0d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:51:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fa0f70db7fa421f82ea7b8f4d46da8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318fb33cab174e079a34e8859e19d8f1 : SUCCESS in 7m 15s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fddf72b24a444da7aaccc61fcc5647ac : FAILURE in 6m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b0e8bf280d49450a8b17da0e09f481ce : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f29c36b09d96c5f2c4e5e84dff9f4492502a9b12","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-05 12:46:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8255132eba77ef4d9f22da0c4b78b74dcdcadf28","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-05 13:35:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"18666de7459541235cbbd5099ef116d7f36d36d9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-05 14:11:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cdd46daf0280b5d6162289152fae5075a64d08cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"date":"2024-07-11 04:18:23.000000000","message":"Abandoned\n\nclose to make monasca more stable","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e18d3e7316c9d74f3cdb5cc6cd740a2dc469fc4","revisions":{"0e645d561058190f968d101c5430b8e4b17f32bd":{"kind":"REWORK","_number":1,"created":"2024-07-04 10:15:28.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/87/923487/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/923487/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/923487/1"}}},"commit":{"parents":[{"commit":"0b682f6d1a676a5fbd92131a60fdc68a295ff77c","subject":"Merge \"Add missing release note links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b682f6d1a676a5fbd92131a60fdc68a295ff77c"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-04 10:14:31.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-04 10:14:31.000000000","tz":420},"subject":"Release monasca for Dalmatian-2 milestone","message":"Release monasca for Dalmatian-2 milestone\n\nChange-Id: I5b77a3e6ee9ee0b91c6107e127415e966ebed23a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e645d561058190f968d101c5430b8e4b17f32bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e645d561058190f968d101c5430b8e4b17f32bd"}]},"branch":"refs/heads/master"},"5f906d97c16d65ef5af31878096eb1dcb9d2adc1":{"kind":"REWORK","_number":2,"created":"2024-07-05 07:57:57.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/87/923487/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/923487/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/923487/2"}}},"commit":{"parents":[{"commit":"0b682f6d1a676a5fbd92131a60fdc68a295ff77c","subject":"Merge \"Add missing release note links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b682f6d1a676a5fbd92131a60fdc68a295ff77c"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-04 10:14:31.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-05 07:57:38.000000000","tz":420},"subject":"Release monasca for Dalmatian-2 milestone","message":"Release monasca for Dalmatian-2 milestone\n\nChange-Id: I5b77a3e6ee9ee0b91c6107e127415e966ebed23a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f906d97c16d65ef5af31878096eb1dcb9d2adc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f906d97c16d65ef5af31878096eb1dcb9d2adc1"}]},"branch":"refs/heads/master"},"0e18d3e7316c9d74f3cdb5cc6cd740a2dc469fc4":{"kind":"REWORK","_number":3,"created":"2024-07-05 10:28:28.000000000","uploader":{"_account_id":13114,"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","username":"thuvh"},"ref":"refs/changes/87/923487/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/87/923487/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/87/923487/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/87/923487/3"}}},"commit":{"parents":[{"commit":"0b682f6d1a676a5fbd92131a60fdc68a295ff77c","subject":"Merge \"Add missing release note links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b682f6d1a676a5fbd92131a60fdc68a295ff77c"}]}],"author":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-04 10:14:31.000000000","tz":420},"committer":{"name":"Hoai-Thu Vuong","email":"thuvh87@gmail.com","date":"2024-07-05 10:28:13.000000000","tz":420},"subject":"Release monasca for Dalmatian-2 milestone","message":"Release monasca for Dalmatian-2 milestone\n\nChange-Id: I5b77a3e6ee9ee0b91c6107e127415e966ebed23a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e18d3e7316c9d74f3cdb5cc6cd740a2dc469fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e18d3e7316c9d74f3cdb5cc6cd740a2dc469fc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
