)]}'
{"id":"openstack%2Freleases~924247","triplet_id":"openstack%2Freleases~master~If51a143448ad8d19f39c781d58b98fb47ac2d9da","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2024-07-26 06:18:20.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2024-07-23 17:52:47.000000000","reason":"\u003cGERRIT_ACCOUNT_6914\u003e replied on the change","reason_account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-23 11:53:22.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-26 06:18:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If51a143448ad8d19f39c781d58b98fb47ac2d9da","subject":"Release Horizon for Dalmatian","status":"MERGED","created":"2024-07-16 15:58:20.000000000","updated":"2024-07-26 06:18:20.000000000","submitted":"2024-07-26 06:18:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924247","meta_rev_id":"34445044f1a0d133230d545c9f79c0ef8e0f8973","_number":924247,"virtual_id_number":924247,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-26 06:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"all":[{"value":1,"date":"2024-07-23 17:52:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2024-07-26 06:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-07-25 11:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2024-07-26 06:11:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-07-23 17:53:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 15:59:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-16 16:09:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 16:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2024-07-16 16:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2024-07-23 10:21:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-25 11:28:54.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"948bd8bdd0f797b691c0b08fa96b1d9d7f0bd712","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-16 15:58:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"776688560960e7a1d3802e191e3b2ef8ffba888c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 16:09:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/820b8d61f9544cbabfe1e424897991d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a898ff36c0648cfa952b1b404e8f561 : SUCCESS in 8m 57s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fa1ffb66e19a45ae99ec8bf9a39adf95 : SUCCESS in 7m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0f8d6215e3b54e5eb1d9c818ceb0bc53 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"25bf14608a50da861a3543844b9b9533247e7704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-16 16:15:23.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a0a26e607c7469279c33ebf287c49155f0e4099f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 16:24:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1d43f41d194e2e91d332a7df428151\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b79a8c71abf46c6942fecc8364369c6 : SUCCESS in 6m 50s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/86902be6462d4662b27954c5ba2aa56f : SUCCESS in 6m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/900bf53ef57d49a8b98c452a2ca6735d : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0b76bd5d2abd163210e59d8749df066f3e62915a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-23 10:21:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af647d6b2baca7127d0e6e04e6a5025f5bd199ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 10:35:32.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"09cd34e0ebda27f121a479a47c6587c0a4a46fa8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 10:36:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb2ba58a4f96ef60d3fc16294209ec91ae27492d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 10:46:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f428a6843334966a95e8f6cee82e814\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b77fd42d476449db7f44998c3abaa1e : SUCCESS in 6m 15s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c617aa3c5ee0457eb550ee18d12b9a56 : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0ebddc89167c406483b909419746f664 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":3},{"id":"dde22a2079595f89d0963ea6b0fa8dc7de5f21cd","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2024-07-23 16:10:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4f2302b871a5142d01c13c90fca96a2f4017754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 16:12:22.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3b40b0c63659c31b47d284951fcd18bcbde9aa03","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-23 16:12:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b2ffc8a6dfbfa3a5d2c9f3274faf9e017a54e4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 16:30:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9633789ab96140249b219721cc36d3ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15a15fa4532415fbd1a89ed50f2a9b1 : SUCCESS in 6m 42s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/5ea3c019761c465781111bfd804facf7 : SUCCESS in 5m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b51cf69bb5be463580a80307fad972a5 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":4},{"id":"f512f17d917e354ee3d5d324e66d50f5458ce590","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2024-07-23 17:52:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e627a30a7705690b397dd7c26a4904ff69e97389","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 17:53:36.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"f2008f54511a835faf729e0f37427adf4a3fb616","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-07-25 11:28:54.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ec10b506ff2a8d61375aacfa40a2b010d94d16ce","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-26 06:11:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c20f9c59963d0c02734d1241ab91e2f7b4cbab44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 06:11:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"afb83248a16e5b4d7e030e4f903b4fe5c076ef98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 06:18:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d4b4a6e238e4b3da91ec95a1287d923\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e92e49b57e49fb903ae364b9bfa1da : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/142f3432808d4361bf00a221b0d7e55b : SUCCESS in 6m 19s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/99552a65706d42a096aeeb0f883483eb : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":4},{"id":"34445044f1a0d133230d545c9f79c0ef8e0f8973","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 06:18:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c65b2d6b168f327f5787b913b1222d77b1e14582","revisions":{"6d8f97a1f964ad0710abb17f7fdd10148e8dd377":{"kind":"REWORK","_number":1,"created":"2024-07-16 15:58:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/924247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/924247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/924247/1"}}},"commit":{"parents":[{"commit":"b7ea66a278aa338672467e7f5dfc346cdd1bb3bc","subject":"Merge \"Mark murano-pkg-check as retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7ea66a278aa338672467e7f5dfc346cdd1bb3bc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-16 15:54:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-16 15:54:46.000000000","tz":540},"subject":"Release Horizon for Dalmatian","message":"Release Horizon for Dalmatian\n\nRecent commit 222f657fbd0b5382ce221b268bd952409cb164a1 in horizon\nchanged data structure obtained by dashboard api clients, and some of\nthe dashboard plugins need to adopt to the new structure.\n\nExample:\nhttps://review.opendev.org/c/openstack/heat-dashboard/+/922761\n\nI\u0027m proposing an early release so that we can properly bump the minimum\nhorizon version, to require the underlying change.\n\nChange-Id: If51a143448ad8d19f39c781d58b98fb47ac2d9da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d8f97a1f964ad0710abb17f7fdd10148e8dd377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d8f97a1f964ad0710abb17f7fdd10148e8dd377"}]},"branch":"refs/heads/master"},"b7cc764184309c8ad91112253b00fa9ebc2a90fc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-16 16:15:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/924247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/924247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/924247/2"}}},"commit":{"parents":[{"commit":"b7ea66a278aa338672467e7f5dfc346cdd1bb3bc","subject":"Merge \"Mark murano-pkg-check as retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7ea66a278aa338672467e7f5dfc346cdd1bb3bc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-16 15:54:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-16 16:15:05.000000000","tz":540},"subject":"Release Horizon for Dalmatian","message":"Release Horizon for Dalmatian\n\nRecent commit 222f657fbd0b5382ce221b268bd952409cb164a1 in horizon\nchanged data structure obtained by neutron client code in horizon, and\nsome of the dashboard plugins need to be updated to adopt to\nthe change.\n\nExample:\nhttps://review.opendev.org/c/openstack/heat-dashboard/+/922761\n\nCreate an early release so that we can properly bump the minimum\nhorizon version, to require the underlying change.\n\nChange-Id: If51a143448ad8d19f39c781d58b98fb47ac2d9da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7cc764184309c8ad91112253b00fa9ebc2a90fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7cc764184309c8ad91112253b00fa9ebc2a90fc"}]},"branch":"refs/heads/master"},"56bced305e40fd092fcd41234561c1f1f9d56288":{"kind":"REWORK","_number":3,"created":"2024-07-23 10:35:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/924247/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/924247/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/924247/3"}}},"commit":{"parents":[{"commit":"b7ea66a278aa338672467e7f5dfc346cdd1bb3bc","subject":"Merge \"Mark murano-pkg-check as retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7ea66a278aa338672467e7f5dfc346cdd1bb3bc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-16 15:54:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 10:35:32.000000000","tz":0},"subject":"Release Horizon for Dalmatian","message":"Release Horizon for Dalmatian\n\nRecent commit 222f657fbd0b5382ce221b268bd952409cb164a1 in horizon\nchanged data structure obtained by neutron client code in horizon, and\nsome of the dashboard plugins need to be updated to adopt to\nthe change.\n\nExample:\nhttps://review.opendev.org/c/openstack/heat-dashboard/+/922761\n\nCreate an early release so that we can properly bump the minimum\nhorizon version, to require the underlying change.\n\nChange-Id: If51a143448ad8d19f39c781d58b98fb47ac2d9da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56bced305e40fd092fcd41234561c1f1f9d56288"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/56bced305e40fd092fcd41234561c1f1f9d56288"}]},"branch":"refs/heads/master"},"c65b2d6b168f327f5787b913b1222d77b1e14582":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-07-23 16:12:22.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/924247/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/924247/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/924247/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/924247/4"}}},"commit":{"parents":[{"commit":"b7ea66a278aa338672467e7f5dfc346cdd1bb3bc","subject":"Merge \"Mark murano-pkg-check as retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7ea66a278aa338672467e7f5dfc346cdd1bb3bc"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-16 15:54:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-07-23 16:12:22.000000000","tz":0},"subject":"Release Horizon for Dalmatian","message":"Release Horizon for Dalmatian\n\nRecent commit a2d84f10e09baaec844386b4fb204b8bfe75ffd0 in horizon\nchanged data structure obtained by neutron client code in horizon, and\nsome of the dashboard plugins need to be updated to adopt to\nthe change.\n\nExample:\nhttps://review.opendev.org/c/openstack/heat-dashboard/+/922761\n\nCreate an early release so that we can properly bump the minimum\nhorizon version, to require the underlying change.\n\nChange-Id: If51a143448ad8d19f39c781d58b98fb47ac2d9da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c65b2d6b168f327f5787b913b1222d77b1e14582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c65b2d6b168f327f5787b913b1222d77b1e14582"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
