)]}'
{"id":"openstack%2Freleases~924299","triplet_id":"openstack%2Freleases~master~I5273705e55ccaaeff43e49e8dc9003e212a4f9f2","project":"openstack/releases","branch":"master","topic":"caracal-stable","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-07-23 09:33:06.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-07-23 09:33:06.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-07-17 14:13:45.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I5273705e55ccaaeff43e49e8dc9003e212a4f9f2","subject":"stable/2024.1: New Neutron release for Caracal 24.0.1","status":"MERGED","created":"2024-07-17 07:02:53.000000000","updated":"2024-07-23 09:33:06.000000000","submitted":"2024-07-23 09:33:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924299-caracal-stable","meta_rev_id":"70ad63b2cae82440987b79423b81e82fc6572665","_number":924299,"virtual_id_number":924299,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 09:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-07-17 07:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-07-19 10:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-07-23 09:25:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2024-07-17 14:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-07-23 09:25:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-07-17 14:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 07:03:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-17 07:12:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-17 07:53:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-07-17 11:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-07-17 11:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-07-19 10:11:08.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-07-23 09:25:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"8080e66770c2d52a468ec5289392a35b55f0a483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-17 07:02:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff1ba7d8d8e6d4ee53058a19d7ad51f3b743f306","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-17 07:03:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"232a27edefb9735ef0fec06347e9a0fd3f6f614b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-17 07:04:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"004c8dc4a5d3e72afe184d86df1171499215defe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 07:12:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c395d424bf5497aa571c1f18d46e267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d14b095809d54af5a5e3c8ddb0ee12d1 : SUCCESS in 7m 00s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/15b182822bb64d4dbd96351d62080b1a : SUCCESS in 6m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/19954b73753642768e35aedde34431c4 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"03da962850bbed07c4939dded5fb25cfedc4fb5a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-07-17 07:53:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4d59b7e74a4f2fcaa5b1983d1becdeeb5706177d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-17 14:13:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93b019efce7225c50dca726328b20dc8a588dc81","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 14:14:32.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"95be3cc177d263d4307d75d4c26696e9e91d15e7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-07-19 10:11:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5c37ddd3665e46e497466de38f92f93e0d3d099a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-23 09:25:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dcc148dc413d0c2b0abea81a7e64967593002a5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:25:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0aa9f7f376fdc6274711ab76d540deb4bf316105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:33:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c90c0b7cfc642e59deef6fa79addb2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c292bd8f678742e6b24c136e2f21f313 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21115cfd8f364342b27952f93cd87eac : SUCCESS in 6m 27s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e999b0312cd8479fbe738f00354f12be : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":2},{"id":"70ad63b2cae82440987b79423b81e82fc6572665","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:33:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"72f52f632534cfbbd3821a9e2f8be36acb448dda","revisions":{"8f3b84973ea1d6c8f3e3f473ae12dfec0615dd26":{"kind":"REWORK","_number":1,"created":"2024-07-17 07:02:53.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/99/924299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/924299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/924299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/924299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/924299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/924299/1"}}},"commit":{"parents":[{"commit":"b7ea66a278aa338672467e7f5dfc346cdd1bb3bc","subject":"Merge \"Mark murano-pkg-check as retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7ea66a278aa338672467e7f5dfc346cdd1bb3bc"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-07-17 06:59:53.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-07-17 06:59:53.000000000","tz":120},"subject":"stable/2024.1: New Neutron release for Caracal 24.0.1","message":"stable/2024.1: New Neutron release for Caracal 24.0.1\n\n$ git log --oneline --no-merges 24.0.0..HEAD\n59bc8e476f Checking pci_slot to avoid changing staus to BUILD forever\n424f1e6f4c [FT] Add a timeout for the NB/SB connection stop method\n94d86ba8b8 Add the port \"fixed_ips\" information in the DHCP RPC\n4c70d3ea61 Change to use selectin for RBACs in SubnetPool DB load strategy\nf25cc2f503 Reorder subnet RBAC policy check strings\n346d433aa8 Return empty BpInfo if missing binding:profile\n3d97c4f2ac [OVN] Sanitize the classless-static-route DHCP option\n7e36795563 [FT] Run test_periodic_sync_routers_task tests serially\nf94f8b6384 Add a default goto table\u003d94 for openvswitch fw\nf0901c75e2 (gerrit/stable/2024.1) [OVN] Bump revision number after update_virtual_port_host\n1098929a54 [OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent\nf7e9b2e5b3 Revert \"Use HasStandardAttributes as parent class for Tags DB model\"\n966fa566e5 Revert \"[OVN] Prevent Trunk creation/deletion with parent port bound\"\n0d8cc09c4a [stable only] Fix KeyError in set_gateway_mtu\n4221f706ce [functional tests] compatibility with ovsdbapp\u003e\u003d2.6.1\n5bdd0efb39 [stable only] Do not fail on missing logical router ports\nd8208fc514 Return both project_id when validating auto allocate network\na61598f64b Put monitors dictionary back in MetadataDriverBase class\nac1472c8cf Don\u0027t update revision number if object was not modified\n51961e195f Fix KeyError failure in _sync_subnet_dhcp_options()\n646270de5b Fixing the 500 HTTP code in the metadata service if Nova is down\n\nChange-Id: I5273705e55ccaaeff43e49e8dc9003e212a4f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f3b84973ea1d6c8f3e3f473ae12dfec0615dd26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f3b84973ea1d6c8f3e3f473ae12dfec0615dd26"}]},"branch":"refs/heads/master"},"72f52f632534cfbbd3821a9e2f8be36acb448dda":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-17 07:04:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/99/924299/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/99/924299/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/924299/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/924299/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/99/924299/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/99/924299/2"}}},"commit":{"parents":[{"commit":"b7ea66a278aa338672467e7f5dfc346cdd1bb3bc","subject":"Merge \"Mark murano-pkg-check as retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7ea66a278aa338672467e7f5dfc346cdd1bb3bc"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-07-17 06:59:53.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-07-17 07:04:29.000000000","tz":0},"subject":"stable/2024.1: New Neutron release for Caracal 24.0.1","message":"stable/2024.1: New Neutron release for Caracal 24.0.1\n\n$ git log --oneline --no-merges 24.0.0..HEAD\n59bc8e476f Checking pci_slot to avoid changing staus to BUILD forever\n424f1e6f4c [FT] Add a timeout for the NB/SB connection stop method\n94d86ba8b8 Add the port \"fixed_ips\" information in the DHCP RPC\n4c70d3ea61 Change to use selectin for RBACs in SubnetPool DB load strategy\nf25cc2f503 Reorder subnet RBAC policy check strings\n346d433aa8 Return empty BpInfo if missing binding:profile\n3d97c4f2ac [OVN] Sanitize the classless-static-route DHCP option\n7e36795563 [FT] Run test_periodic_sync_routers_task tests serially\nf94f8b6384 Add a default goto table\u003d94 for openvswitch fw\nf0901c75e2 [OVN] Bump revision number after update_virtual_port_host\n1098929a54 [OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent\nf7e9b2e5b3 Revert \"Use HasStandardAttributes as parent class for Tags DB model\"\n966fa566e5 Revert \"[OVN] Prevent Trunk creation/deletion with parent port bound\"\n0d8cc09c4a [stable only] Fix KeyError in set_gateway_mtu\n4221f706ce [functional tests] compatibility with ovsdbapp\u003e\u003d2.6.1\n5bdd0efb39 [stable only] Do not fail on missing logical router ports\nd8208fc514 Return both project_id when validating auto allocate network\na61598f64b Put monitors dictionary back in MetadataDriverBase class\nac1472c8cf Don\u0027t update revision number if object was not modified\n51961e195f Fix KeyError failure in _sync_subnet_dhcp_options()\n646270de5b Fixing the 500 HTTP code in the metadata service if Nova is down\n\nChange-Id: I5273705e55ccaaeff43e49e8dc9003e212a4f9f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72f52f632534cfbbd3821a9e2f8be36acb448dda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/72f52f632534cfbbd3821a9e2f8be36acb448dda"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
