)]}'
{"id":"openstack%2Freleases~924414","triplet_id":"openstack%2Freleases~master~I3ed7a99fdd0bb9b0351578467c632335b4e90db1","project":"openstack/releases","branch":"master","topic":"neutron_stable_releases","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-07-23 09:33:58.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-07-23 09:33:58.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-07-18 20:34:42.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I3ed7a99fdd0bb9b0351578467c632335b4e90db1","subject":"stable/2023.1: New Neutron release for Antelope 22.2.0","status":"MERGED","created":"2024-07-18 15:07:14.000000000","updated":"2024-07-23 09:33:58.000000000","submitted":"2024-07-23 09:33:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924414-neutron_stable_releases","meta_rev_id":"037ae9f461d9213f8b5bc67f71ce49746db2e86c","_number":924414,"virtual_id_number":924414,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-23 09:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-07-19 10:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-07-23 09:25:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2024-07-18 20:34:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-07-23 09:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-07-18 20:35:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-18 15:08:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-18 15:20:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-07-18 15:20:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-07-18 15:24:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-19 10:12:24.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-07-23 09:25:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"d5d91d434dfc04f4f909051355c12c000e122d7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-18 15:07:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e14135fe7b2956372b9e22d47098e8d101994993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 15:24:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f4a0e46d3c14d038a2b6be1d7534d98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e5d70133adf426a8a21ec627a4c0e34 : SUCCESS in 6m 22s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d67c8d042f574331b113b66dfbc4e7fd : SUCCESS in 6m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/21ad5052d2b046588306d43bafa4a34d : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e016a20a6b61969728d31f7285a856c31e122252","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-07-18 20:34:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a988cf45f970550d0c73c21f657de28accade99","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 20:35:30.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"c909309c4c235e61a07b2e13fb3bd7dd94c6395e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-07-19 10:12:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e23e1bc31bd0cdf3695e2e03c58e71bf94501a0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-07-23 09:25:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8ee5a3a91aab969a6cd51ae76d7021481b1133e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:26:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f19f7052188806517ad5f0dd236a7a174c56c30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:33:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea4795fa1494a6eb7ff1f748542ef19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0eafb9ca564457ba0d82ac3cbc35630 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49420e74f7f4453e932e0b8769b013e3 : SUCCESS in 6m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e52cffefbb0848c8883785a35e56658b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"037ae9f461d9213f8b5bc67f71ce49746db2e86c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:33:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1fd6d3555149400160e33a966a0ed206c9a1a2aa","revisions":{"1fd6d3555149400160e33a966a0ed206c9a1a2aa":{"kind":"REWORK","_number":1,"created":"2024-07-18 15:07:14.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/14/924414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/924414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/924414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/924414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/924414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/924414/1"}}},"commit":{"parents":[{"commit":"b7ea66a278aa338672467e7f5dfc346cdd1bb3bc","subject":"Merge \"Mark murano-pkg-check as retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b7ea66a278aa338672467e7f5dfc346cdd1bb3bc"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-07-18 15:05:17.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-07-18 15:05:17.000000000","tz":120},"subject":"stable/2023.1: New Neutron release for Antelope 22.2.0","message":"stable/2023.1: New Neutron release for Antelope 22.2.0\n\n$ git log --oneline --no-merges 22.1.0..HEAD\n27eee0b9e8 Checking pci_slot to avoid changing staus to BUILD forever\n603f388ae9 [FT] Add a timeout for the NB/SB connection stop method\n182b5acbdf Add the port \"fixed_ips\" information in the DHCP RPC\n46dbdcf3f2 Change to use selectin for RBACs in SubnetPool DB load strategy\n14dca0116e Return empty BpInfo if missing binding:profile\nd32fa358c5 Increase timeout in test_get_all_devices()\n65be5a940b Add debug information to ``MacvtapAgentTestCase.test_get_all_devices``\n9de68a7408 Cleanup before executing \"test_get_all_devices\"\n498eaa16a0 [2023.1 Only][OVN] Add the bridge name and datapath type to the port VIF details\n0fb74dbbe1 [OVN] Sanitize the classless-static-route DHCP option\n114f68fa59 [FT] Run test_periodic_sync_routers_task tests serially\n1045985c76 Add a default goto table\u003d94 for openvswitch fw\n27869542aa [OVN] Bump revision number after update_virtual_port_host\ncbcfa692fc [OVN] Fix virtual parent match for PortBindingUpdateVirtualPortsEvent\nebd52b0df3 [OVN] Update lsp host id when virtual parent moves\n4f5c023279 Revert \"Use HasStandardAttributes as parent class for Tags DB model\"\n961a7a304c [ML2/OVN] Add gateway_port support for FIP\ncd92d42cfa [stable only] Fix KeyError in set_gateway_mtu\nfde31034ae [functional tests] compatibility with ovsdbapp\u003e\u003d2.2.2\nda8f6ef4dd [stable only] Do not fail on missing logical router ports\n75eefdbc69 Return both project_id when validating auto allocate network\nc190408171 [stable/2023.1 only] Fix missing test fix for ext gw\n0fe6b7cec0 Don\u0027t update revision number if object was not modified\n2ec0ef0909 Fix KeyError failure in _sync_subnet_dhcp_options()\n872c6ad2cd Enhance IptablesFirewallDriver with remote address groups\n0a421118f2 [OVN] The L3 scheduler does not use all chassis by default\n9bb37523a5 [stable/2023.1 only] Fix incorrect cherry-pick for ext gw\na8bf8cdaa6 Fix TestOVNMechanismDriver ipv6 tests\n10af328885 Use the system-dependent string for IP protocol 4\nd17b1a6abb Fix iptables mapping of \u0027ipip\u0027 protocol\n5ffec5d17a [OVN] Set MTU of the VETH interfaces between OVS and metadata\n0673a9bb23 [stable-only][OVN] Set VETH interface MAC address before up\n0dfe8dedd6 Ensure that haproxy spawned by the metadata agents is active\nb34911a6c9 [OVN] A LRP in an external tunnelled network has no chassis\n1d2a1c68c0 Retry ``set|get_link_attribute(s)`` if the interface is not present\n475a25259f [Docs] Add info about incompatible PF configuration in ML2/OVN gaps\nbd6595ddb8 Log warning about port forwardings that won\u0027t work properly\nc40f637bde [OVN] Warn about invalid OVN and FIP PF config during start of Neutron\n45acc0c53b Disallow subnet cidr of :: without PD\nc019a42086 [OVN] Add ``subnetpool-prefix-ops`` extension to ML2/OVN mech driver\n331f6ae240 Unmaintained-Only: change stable/yoga to unmaintained\n3dec91ec58 Change SG rules backref load method to \"joined\"\n68542c04a1 [OVN] Use elevated context to retrieve subnet in router port configuration\nfd0eabcf36 dhcp: fix usage of helper function to retrieve process name\n2f678b980c [OVN][FT] Retry in case of timeout when executing \"ovsdb-client\".\nf8628b5c04 [Fullstack] Consolidate segmentation_id update tests into single test\ne5c743b820 [Fullstack] Remove SecurityGroupRulesTest.test_normalized_cidr_in_rule test\n37367c1092 [Fullstack] Remove test_port_shut_down module\n1226d6adcd [Fullstack] Remove unnecessary tests\n17d15db7f4 Set trunk parent port as access port in ovs to avoid loop\n09c1c61677 Make get_ports RPC method common for the DHCP and Metadata agent\n20ba447787 Make ``OVNMechanismDriver.post_fork_initialize`` callback cancellable\n3eddcd970d If method ``set_netns`` fails, restore previous device namespace\n643dbbbf6b [OVN] OVN agent extensions correctly consume agent API\na991b097c1 [OVN] Retrieve the OVN agent extensions correctly\n5ff38217c7 Mark \"test_port_creation_and_deletion\" as unstable\na72b44a3a9 Forbid the subnet gateway IP deletion if a router interface is attached\nbd1927028e Register Chassis_Private table in BaseOvnSbIdl\nc5e70ad716 Add sleep before checking if ovs port is in the namespace\n69c49c4ef2 fix netns deletion of broken namespaces\n94cf7a4c28 Add max limit to agent_down_time\n696b2f4d4b Increase port name size and type to internal\nb992d639b9 Handle creation of Port_Binding with chassis set\n63c16672a5 Metadata: handle process exceptions\n11b977ca39 [UT] OVN fake resources factory method should return instance\n29d5570ab3 Improve the SG RPC callback ``security_group_info_for_ports``\n1d7d489b3a [ovn][ipv6] Add some more tests to skiplist\na7cff2d286 [ovn][ipv6] Skip test_update_router_admin_state\n0aeb703db7 Make unit tests compatible with neutron-lib 3.4.2\nb9ffa5aa13 [OVN] Add update event to ``OVSInterfaceEvent`` class\n7c7a99acb8 Remove obsolete PID files before start\n6205158831 ovn-metadata: Refactor events\n8f0ac0a201 Update the External Ports documentation\n1dbe4248bb Enhanced external port scheduling\nd69d8688b8 Reintroduce agent bridge resync test\nc4b322597b [OVN][FT] Make explicit the \"publish\" call check in \"test_port_forwarding\"\ncc642431f0 Fix the common/ovn functional tests\n8c26736027 Improve \"sync_ha_chassis_group\" method\n877e85e5a8 [OVN] DB sync host/physnet - filter on agent_type\nf2ec3a6cec Correctly validate subnet arguments when using a subnetpool\na28ec9ed3c [Fullstack] Double check that agent is dead when it should be dead\nfd06c73665 For hosts in DVR mode, only fetch bound FIPs\nae90f380a8 [DHCP agent] Fetch OVN Metadata port from plugin\nd235f65516 get_hosts_mapped_with_segments add filter agt_type\nf81549c0e5 [DHCP agent] Fix route to OVN metadata port for non-isolated networks\nde0755232f Add constant to identify OVN LB HM ports\n458915e56f docs: update default value of metadata workers for ml2/ovn\n87f7b9a46c Catch non-existent entry failures better in ip_lib\nba8335375a Don\u0027t set port capabilities for OVS HW offloading\n37bdb9e7d5 Add scope ID to the \"GROUP BY\" clause in ``get_scoped_floating_ips``\nf96691b208 Remove duplicate rows in MySQL query output\n1514fbe1f8 [OVN] Set the Neutron port status based on \"lsp.up\" and \"lsp.enabled\"\n0de8e70d29 [OVN] Add a log message after the \"post_fork_initialize\" method\n\nChange-Id: I3ed7a99fdd0bb9b0351578467c632335b4e90db1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fd6d3555149400160e33a966a0ed206c9a1a2aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fd6d3555149400160e33a966a0ed206c9a1a2aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
