)]}'
{"id":"openstack%2Freleases~926515","triplet_id":"openstack%2Freleases~master~I0558bb05c35a88c69db79326123b542c155533ba","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-21 14:58:06.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-08-21 14:44:18.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-21 14:58:06.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-08-21 14:58:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0558bb05c35a88c69db79326123b542c155533ba","subject":"Release oslo.context 5.6.0 for 2024.2 dalmatian.","status":"MERGED","created":"2024-08-19 09:42:40.000000000","updated":"2024-08-21 14:58:06.000000000","submitted":"2024-08-21 14:58:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926515","meta_rev_id":"e93d85363efd2f51b66e61dbfffe54c3b6818711","_number":926515,"virtual_id_number":926515,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 14:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-21 14:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-21 14:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-21 14:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-21 14:40:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2024-08-21 14:58:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 09:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 11:20:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-21 09:41:18.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-21 14:40:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"edfbeb65fa957ba4698682871349d827bd34ae85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-19 09:42:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"997c6f97dacd67925a45589472a4196e77a84976","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 09:43:28.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"ff895ec2a7da6bfdc7f7b7e47b5824f5ea5cb52d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-19 09:51:45.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":2},{"id":"5db45ed8b8f69f0bddc2fe16a6166b5eae72c2f8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 09:53:26.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"8be2c5964374e0996920827929ac7dd452e30ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 10:05:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2f2ec19793b4b2891a8234abd143a20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b43f1c53294410ad1d793f57433a81 : SUCCESS in 6m 43s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/0db79779ce034cffbf5c92e7ce04de72 : SUCCESS in 5m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b463b4321dd74b9d9ce4fa050c4d3740 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1eda175bfb4020e647f4c416239f49ebe4d8d986","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-20 14:42:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f334a51cc2611f22169e5187e421b2ceb3661d9c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-21 09:41:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cc71d67d067fd794d127e1242884736a4497645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:39:59.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"35901f9e9a6edd34048c319dbacf8786f85f1be1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:40:50.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e799bbe0c3043f8c448419db562dcc2b33d35b2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:40:53.000000000","message":"Patch Set 3: -PTL-Approved","accounts_in_message":[],"_revision_number":3},{"id":"60830d3c401edc6e24a94feec8014bfc304b5357","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-21 14:44:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"474294b874f8844e7c97ecbc321ebb9b6241a964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:48:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b89ae93549bb4b35aee842fa8b424663\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4bfb49a46d4201aef7c32f6662feba : SUCCESS in 5m 49s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6646aa9fa8c44cce995d4bc2f6e52690 : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cac2ba17d6bc429c9745413a32737a31 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"46c090a4d3761e130ba97c1430c546e0d5fd5e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:48:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe13ecd5dbf6482344cb5a578d709a0055a0c677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:58:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1bcd670ae464db19aa378fce5a87539\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0422751d0ae84b5b87f5ef552b9f703c : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07380eed1f074f09bb93453f2e96efd4 : SUCCESS in 7m 57s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/4043495c574f4e37bd4c03d4da2db171 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":3},{"id":"e93d85363efd2f51b66e61dbfffe54c3b6818711","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:58:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5f10f1e5daaf94cab16284e6ec97d2683582cfd4","revisions":{"adc43d3f60f04febce11eec963252a92ae4a06bd":{"kind":"REWORK","_number":1,"created":"2024-08-19 09:42:40.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/15/926515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/926515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/926515/1"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 09:40:42.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 09:40:42.000000000","tz":120},"subject":"Release oslo.concurrency 5.5.1 for 2024.2 dalmatian.","message":"Release oslo.concurrency 5.5.1 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 5.5.0..7f3b6acb7f006dd156a51882be533c17721b878b\n\n292349d reno: Update master for unmaintained/zed\n06b54af Remove old excludes\n37ffad4 pre-commit: Remove outdated comment\n9568f54 Update master for stable/2024.1\nf47d6f4 reno: Update master for unmaintained/xena\n6e89ad9 reno: Update master for unmaintained/wallaby\nd71d6fa reno: Update master for unmaintained/victoria\n\nChange-Id: I0558bb05c35a88c69db79326123b542c155533ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adc43d3f60f04febce11eec963252a92ae4a06bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/adc43d3f60f04febce11eec963252a92ae4a06bd"}]},"branch":"refs/heads/master"},"23866881c368ff799b400139a7e0aff7cc2f64e4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-19 09:51:45.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/15/926515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/926515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/926515/2"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 09:40:42.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 09:51:15.000000000","tz":120},"subject":"Release oslo.context 5.5.1 for 2024.2 dalmatian.","message":"Release oslo.context 5.5.1 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 5.5.0..7f3b6acb7f006dd156a51882be533c17721b878b\n\n292349d reno: Update master for unmaintained/zed\n06b54af Remove old excludes\n37ffad4 pre-commit: Remove outdated comment\n9568f54 Update master for stable/2024.1\nf47d6f4 reno: Update master for unmaintained/xena\n6e89ad9 reno: Update master for unmaintained/wallaby\nd71d6fa reno: Update master for unmaintained/victoria\n\nChange-Id: I0558bb05c35a88c69db79326123b542c155533ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23866881c368ff799b400139a7e0aff7cc2f64e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23866881c368ff799b400139a7e0aff7cc2f64e4"}]},"branch":"refs/heads/master"},"5f10f1e5daaf94cab16284e6ec97d2683582cfd4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-08-21 14:39:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/15/926515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/926515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/926515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/926515/3"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 09:40:42.000000000","tz":120},"committer":{"name":"Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-08-21 14:39:59.000000000","tz":0},"subject":"Release oslo.context 5.6.0 for 2024.2 dalmatian.","message":"Release oslo.context 5.6.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 5.5.0..7f3b6acb7f006dd156a51882be533c17721b878b\n\n292349d reno: Update master for unmaintained/zed\n06b54af Remove old excludes\n37ffad4 pre-commit: Remove outdated comment\n9568f54 Update master for stable/2024.1\nf47d6f4 reno: Update master for unmaintained/xena\n6e89ad9 reno: Update master for unmaintained/wallaby\nd71d6fa reno: Update master for unmaintained/victoria\n\nChange-Id: I0558bb05c35a88c69db79326123b542c155533ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f10f1e5daaf94cab16284e6ec97d2683582cfd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5f10f1e5daaf94cab16284e6ec97d2683582cfd4"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
