)]}'
{"id":"openstack%2Freleases~926534","triplet_id":"openstack%2Freleases~master~I94daa57bb89b0a47d7371e66918d07b8a77c3f48","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-20 14:41:29.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-08-21 15:11:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94daa57bb89b0a47d7371e66918d07b8a77c3f48","subject":"Release oslo.rootwrap 7.3.0 for 2024.2 dalmatian.","status":"MERGED","created":"2024-08-19 12:58:19.000000000","updated":"2024-08-21 15:11:49.000000000","submitted":"2024-08-21 15:11:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926534","meta_rev_id":"d5ea057f12acc0d345e686fc9150bea2d133e985","_number":926534,"virtual_id_number":926534,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 15:11:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-21 14:46:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-21 14:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-08-21 14:46:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-21 14:46:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-08-21 14:47:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 12:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 13:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-21 14:23:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-21 14:46:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"e78c0795c1ff91d51a09c3aa16ac4e3ae9eee243","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-19 12:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f75b6e0668f27aafceb423b1ec32a15ec7f20e70","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 12:59:32.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"ab866fabf6dd1709afd3c0622c41918fba99377c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:08:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1549753c24d44e9e96f44312d341999f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049b944016ea440eb72b1b94988f75c7 : SUCCESS in 6m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/84b21b6fdac943dfb93d66612bc206a6 : SUCCESS in 4m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3555d85d91324e958cd8ecb1605aa658 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9e92d5429604f58bec9ebf04113432ce6a2e9ed5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-20 14:41:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a3c91e181f186912b0fb549273f707e5d53a0298","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-21 14:23:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6f081f17472f7990cced220c013a60b4c007584","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:46:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_31245\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"_revision_number":2},{"id":"d5972dec2f9902d933d6e84509e74691029672b6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:46:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4ee8f5729fa4205cd3070688a896e4ce787fd4e8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:47:35.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9a9c75d53f4244253beb45c63517199c47db2def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:02:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a38106890e24450cb0d4a4dd7a50ec60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeba566babde4dfbad50c5cfb671d74a : SUCCESS in 7m 12s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d07e1a8434a142e4b1273aaf8a3700fe : SUCCESS in 5m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6af97bbeb441464cb339af022f2bb98f : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"78734969ebc094cf5bdfc0f5854e742570441b1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:03:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d32020d5c1c065e7a1eda9aa8f475ca5bc346e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:11:49.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c71d4d683f544ca1a604ea13a50ff324\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16cfaa6a06bc40b792640a014381c62f : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fa3fa3704544bc897df626fe93d460 : SUCCESS in 5m 48s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1d2935980b2a4ab7b4097f60548af3b1 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":2},{"id":"d5ea057f12acc0d345e686fc9150bea2d133e985","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:11:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"782bd0cb40cbcfd5c6d00cd8ad74702da4a040e2","revisions":{"ae8d22957e8903497991133ecccc2741922fe1b5":{"kind":"REWORK","_number":1,"created":"2024-08-19 12:58:19.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/34/926534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/926534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/926534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/926534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/926534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/926534/1"}}},"commit":{"parents":[{"commit":"db96ea0f025e15cad9c7a9053fd83b3d4b0315f9","subject":"Release oslo.reports 3.4.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/db96ea0f025e15cad9c7a9053fd83b3d4b0315f9"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 12:55:38.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 12:55:38.000000000","tz":120},"subject":"Release oslo.rootwrap 7.3.0 for 2024.2 dalmatian.","message":"Release oslo.rootwrap 7.3.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 7.2.0..c42ef394dd70503fc0b12a6a2d84f1173fd2bcd7\n\nc42ef39 (HEAD -\u003e master, origin/master, origin/HEAD) Remove unused reno from test requirements\ne6653b5 Remove old excludes\n4b1f916 Update master for stable/2024.1\n3ed850a reno: Update master for unmaintained/victoria\n\nChange-Id: I94daa57bb89b0a47d7371e66918d07b8a77c3f48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae8d22957e8903497991133ecccc2741922fe1b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae8d22957e8903497991133ecccc2741922fe1b5"}]},"branch":"refs/heads/master"},"782bd0cb40cbcfd5c6d00cd8ad74702da4a040e2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-21 14:46:34.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/34/926534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/34/926534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/926534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/926534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/34/926534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/34/926534/2"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 12:55:38.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-08-21 14:46:34.000000000","tz":0},"subject":"Release oslo.rootwrap 7.3.0 for 2024.2 dalmatian.","message":"Release oslo.rootwrap 7.3.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 7.2.0..c42ef394dd70503fc0b12a6a2d84f1173fd2bcd7\n\nc42ef39 (HEAD -\u003e master, origin/master, origin/HEAD) Remove unused reno from test requirements\ne6653b5 Remove old excludes\n4b1f916 Update master for stable/2024.1\n3ed850a reno: Update master for unmaintained/victoria\n\nChange-Id: I94daa57bb89b0a47d7371e66918d07b8a77c3f48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/782bd0cb40cbcfd5c6d00cd8ad74702da4a040e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/782bd0cb40cbcfd5c6d00cd8ad74702da4a040e2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
