)]}'
{"id":"openstack%2Freleases~926535","triplet_id":"openstack%2Freleases~master~I6699c43370c6253f3c04e9dc804dd77cddbfa3b6","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-08-21 14:25:47.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-21 15:07:45.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-08-21 15:07:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6699c43370c6253f3c04e9dc804dd77cddbfa3b6","subject":"Release oslo.upgradecheck 2.4.0 for 2024.2 dalmatian.","status":"MERGED","created":"2024-08-19 13:33:01.000000000","updated":"2024-08-21 15:07:45.000000000","submitted":"2024-08-21 15:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926535","meta_rev_id":"ff72b92e77b72490e34552aab9d0413bff8e06ec","_number":926535,"virtual_id_number":926535,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 15:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-21 14:47:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-21 14:47:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-21 14:47:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-08-21 14:48:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 13:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 14:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-21 10:16:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-21 14:47:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"2d37a675528cdbbf09627f345d1d8ca1a8f11551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-19 13:33:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b00719fe5ec568798cab656adfff730691a0aab6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:34:15.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"6ce569ebd974249295f085656be70fa2ec51c290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:40:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc1e75ce8f5145e4b84182f130cc3294\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cdc63fd0fc54e25a23b994cf87a3a60 : SUCCESS in 6m 57s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/90cb48b874814359b47957298c593b05 : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b365f260358f43cb92f9449a8eb3e83e : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a9a320e1db792f074bb8f6b9aaaa787ef71d7bc2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-20 14:40:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"548a79e9946f5b34900126823b3dd2a0229a953a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-21 10:16:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6271fa70ed2a2440524d3d87ab00be47f19576f4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-21 10:17:40.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64e21cf8c1b9d164109ed59be91a7fc2caed136b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-21 10:53:11.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ea0cc97a011c2e3f616600f8ad2fadb288518baf","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-21 10:53:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6093cd9d251ab2ade5742cd888c062bce26be2f","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-21 10:54:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf1c3b1ca5d1b509830c66d57938dafa6cd67ff7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 10:54:27.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"056f56acbe4dd2b7a10bb814313d7cb7a6726ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 11:01:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcd244c0d06240f3b126da54705e627b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee0f2d52ef654c6ba327b6c1b1df8377 : SUCCESS in 5m 46s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/9a73e1e4f2e042b39e2216bde1e0fbb3 : SUCCESS in 4m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e563ba5c14fb44ad9180e81ad32dacdb : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"08f35b60662f1178091be2a502b11bf8c5177f81","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-21 14:25:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f08008e823f3e8285811051cbf827d1ed0aa583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:47:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_31245\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"_revision_number":3},{"id":"7a78abc3eada73b22d1b60b9a80df9d8ccb12bee","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:47:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"74f83084a0428d5da677acbbac8ffb9bb41c1f12","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:48:24.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"7a578a921e844b80a5aceb81a078f27a9808524d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:56:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bef09e846ae94783a96b707fd70995fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d9c737fcc5423fa123c4a23ed261ce : SUCCESS in 6m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/61d265b0410246ff94cdcbaa1849fefe : SUCCESS in 6m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a00eb958e664496eb39a3c5374642511 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"08c70e8d3e9a5972452b65ff2a2b925c30969550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:57:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"360f550ea42952ed53ab6cf0e229f0b273e7f43c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:07:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d835b440b1fe4cddb945906ac0a588e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504e15af79db4326892bb36961535d43 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35802d24a2747deb26984069d588484 : SUCCESS in 6m 16s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6bd1ef934b3348d1a28d2d281e048b14 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ff72b92e77b72490e34552aab9d0413bff8e06ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:07:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"173d714524209f5d8467d366e44dcd155b08f934","revisions":{"6ee0c5891966f726477fc24b1904933d194f8860":{"kind":"REWORK","_number":1,"created":"2024-08-19 13:33:01.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/35/926535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/926535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/926535/1"}}},"commit":{"parents":[{"commit":"ae8d22957e8903497991133ecccc2741922fe1b5","subject":"Release oslo.rootwrap 7.3.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae8d22957e8903497991133ecccc2741922fe1b5"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 13:32:36.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 13:32:36.000000000","tz":120},"subject":"Release oslo.rootwrap 2.4.0 for 2024.2 dalmatian.","message":"Release oslo.rootwrap 2.4.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 2.3.0..2e41b6df4aa5311fdb088a5c245788a701219072\n\n2e41b6d (HEAD -\u003e master, origin/master, origin/HEAD) reno: Update master for unmaintained/zed\nd4cb382 Remove old excludes\ne3be11c Update master for stable/2024.1\n54ac9b5 reno: Update master for unmaintained/xena\n2dd6705 reno: Update master for unmaintained/wallaby\nd9e8b76 reno: Update master for unmaintained/victoria\n\nChange-Id: I6699c43370c6253f3c04e9dc804dd77cddbfa3b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ee0c5891966f726477fc24b1904933d194f8860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6ee0c5891966f726477fc24b1904933d194f8860"}]},"branch":"refs/heads/master"},"6f2113eb7160167ab529667200f7a844844358cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-21 10:53:11.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/35/926535/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/926535/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/926535/2"}}},"commit":{"parents":[{"commit":"ae8d22957e8903497991133ecccc2741922fe1b5","subject":"Release oslo.rootwrap 7.3.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae8d22957e8903497991133ecccc2741922fe1b5"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 13:32:36.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-21 10:52:53.000000000","tz":120},"subject":"Release oslo.upgradecheck 2.4.0 for 2024.2 dalmatian.","message":"Release oslo.upgradecheck 2.4.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 2.3.0..2e41b6df4aa5311fdb088a5c245788a701219072\n\n2e41b6d (HEAD -\u003e master, origin/master, origin/HEAD) reno: Update master for unmaintained/zed\nd4cb382 Remove old excludes\ne3be11c Update master for stable/2024.1\n54ac9b5 reno: Update master for unmaintained/xena\n2dd6705 reno: Update master for unmaintained/wallaby\nd9e8b76 reno: Update master for unmaintained/victoria\n\nChange-Id: I6699c43370c6253f3c04e9dc804dd77cddbfa3b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f2113eb7160167ab529667200f7a844844358cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f2113eb7160167ab529667200f7a844844358cf"}]},"branch":"refs/heads/master"},"173d714524209f5d8467d366e44dcd155b08f934":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-21 14:47:34.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/926535/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/35/926535/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/35/926535/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/35/926535/3"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 13:32:36.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-08-21 14:47:34.000000000","tz":0},"subject":"Release oslo.upgradecheck 2.4.0 for 2024.2 dalmatian.","message":"Release oslo.upgradecheck 2.4.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 2.3.0..2e41b6df4aa5311fdb088a5c245788a701219072\n\n2e41b6d (HEAD -\u003e master, origin/master, origin/HEAD) reno: Update master for unmaintained/zed\nd4cb382 Remove old excludes\ne3be11c Update master for stable/2024.1\n54ac9b5 reno: Update master for unmaintained/xena\n2dd6705 reno: Update master for unmaintained/wallaby\nd9e8b76 reno: Update master for unmaintained/victoria\n\nChange-Id: I6699c43370c6253f3c04e9dc804dd77cddbfa3b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/173d714524209f5d8467d366e44dcd155b08f934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/173d714524209f5d8467d366e44dcd155b08f934"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
