)]}'
{"id":"openstack%2Freleases~926544","triplet_id":"openstack%2Freleases~master~Ib558718fc4cfba642caacd299eae242fa12749c3","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-20 14:40:37.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-08-22 08:58:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib558718fc4cfba642caacd299eae242fa12749c3","subject":"Release oslo.config 9.6.0 for 2024.2 dalmatian.","status":"MERGED","created":"2024-08-19 15:32:11.000000000","updated":"2024-08-22 08:58:12.000000000","submitted":"2024-08-22 08:58:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926544","meta_rev_id":"d3d4c6dc07b8a6ecee4d6e2791a8f87be868f97a","_number":926544,"virtual_id_number":926544,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-22 08:58:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-21 14:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-22 08:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-08-21 14:50:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-22 08:48:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-08-21 14:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 15:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 16:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-21 14:50:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-22 08:48:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b591375f0d146dc720e12161b65aef01b01f69ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-19 15:32:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"119d3cabde4536b52fad372323c036c73eff162e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 15:33:24.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"5a2a96b31db4474e19336d903c9d2b77901990c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 15:42:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d3201a7d0f402989e949578e9a80a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/534dc7f88a47496e9c99c1081e96f243 : SUCCESS in 9m 01s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/720d7ca6bb1e45feb1807e21542c537d : SUCCESS in 4m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/aa40bac83e9d45448744ce560d404f7a : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"182c1b65d27d30c24c8359a43a9641b27203e627","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-20 14:40:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ba807ac9d2b7853e496dd7d03f1e82fcfd8dc93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-21 10:53:11.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"917d0e37796d394772bd63635677d6269cf3b9cf","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 10:54:37.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"ce5f2b6e0a1a99845caa028f85fe96eeab05eb65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 11:02:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff9d048795444624b680964f86b72f53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50bd766f93f4af1bab4c0c138b68627 : SUCCESS in 8m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/05bcc3f3cafc4f8b8ac99863e9d212a7 : SUCCESS in 6m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a6ddee5c7b0c413cb47a98f4d88ab9ca : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"68c5c2915afa15f8f55f876202e46709a5baa9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:50:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_31245\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"_revision_number":3},{"id":"dc2293c55d69313526b90db780514c47ddb3e75f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:50:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"78bc72262269e4ff64448c5834527914430cfb91","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:51:29.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"cbc9b5d39a387d81887493e4d6bd19d23e0418c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:03:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41a0a29136b64063ab8d87a68a760a8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf99e25bfd045ac8147f8d2c79b4c6c : SUCCESS in 7m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/eb79481ebc064c29a1a5bf00af485315 : SUCCESS in 4m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a45a4dc1c56e4591b6d6444a0d183d76 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":3},{"id":"45ce6ca91d8e0fd5625f8321cceb849b15fe7635","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-22 08:48:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3749667ada465435c604b1558a811ffc668e2dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 08:49:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"07bb2438512bf710958e9e520b04776404397b20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 08:58:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c99861b640493dba980a833d163fb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b30b9b9558848f99f1251a2b9550b69 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5bcd2be0df4890baeebf6fc1ea2a70 : SUCCESS in 7m 06s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c9644aa34fba455bade482ee4cb7a623 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"d3d4c6dc07b8a6ecee4d6e2791a8f87be868f97a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 08:58:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"22cb5f795df6fb6cecdbb0ad3387941b076260e7","revisions":{"65ed0c311e29276625a239a6df762fad25e8a8db":{"kind":"REWORK","_number":1,"created":"2024-08-19 15:32:11.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/44/926544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/926544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/926544/1"}}},"commit":{"parents":[{"commit":"3b0ac9ff009a5cb30eb58572c7db5421a322e90d","subject":"Release oslo.vmware 4.5.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b0ac9ff009a5cb30eb58572c7db5421a322e90d"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 15:31:56.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 15:31:56.000000000","tz":120},"subject":"Release oslo.config 9.6.0 for 2024.2 dalmatian.","message":"Release oslo.config 9.6.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 9.5.0..beb7b6ce2b8a36376324c300328e678bd02ae63a\n\n033eb03 Provide an exit code when raising SystemExit.\n7f4e23b Add option for generate shell completion script\n\nChange-Id: Ib558718fc4cfba642caacd299eae242fa12749c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65ed0c311e29276625a239a6df762fad25e8a8db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65ed0c311e29276625a239a6df762fad25e8a8db"}]},"branch":"refs/heads/master"},"238cb9ff9aa4a433773bde22f8cef932ea1d4363":{"kind":"NO_CHANGE","_number":2,"created":"2024-08-21 10:53:11.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/44/926544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/926544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/926544/2"}}},"commit":{"parents":[{"commit":"015055852f52c2274979ab53d0e5d167479e9da4","subject":"Release oslo.vmware 4.5.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/015055852f52c2274979ab53d0e5d167479e9da4"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 15:31:56.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-21 10:52:54.000000000","tz":120},"subject":"Release oslo.config 9.6.0 for 2024.2 dalmatian.","message":"Release oslo.config 9.6.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 9.5.0..beb7b6ce2b8a36376324c300328e678bd02ae63a\n\n033eb03 Provide an exit code when raising SystemExit.\n7f4e23b Add option for generate shell completion script\n\nChange-Id: Ib558718fc4cfba642caacd299eae242fa12749c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/238cb9ff9aa4a433773bde22f8cef932ea1d4363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/238cb9ff9aa4a433773bde22f8cef932ea1d4363"}]},"branch":"refs/heads/master"},"22cb5f795df6fb6cecdbb0ad3387941b076260e7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-21 14:50:38.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/44/926544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/926544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/926544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/926544/3"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-19 15:31:56.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-08-21 14:50:38.000000000","tz":0},"subject":"Release oslo.config 9.6.0 for 2024.2 dalmatian.","message":"Release oslo.config 9.6.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 9.5.0..beb7b6ce2b8a36376324c300328e678bd02ae63a\n\n033eb03 Provide an exit code when raising SystemExit.\n7f4e23b Add option for generate shell completion script\n\nChange-Id: Ib558718fc4cfba642caacd299eae242fa12749c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22cb5f795df6fb6cecdbb0ad3387941b076260e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22cb5f795df6fb6cecdbb0ad3387941b076260e7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
