)]}'
{"id":"openstack%2Freleases~926622","triplet_id":"openstack%2Freleases~master~I8712e4d14c037c7bcb21bbf0f02e46d44a9b0629","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-22 08:44:54.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-22 02:48:41.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-08-22 11:43:29.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2024-08-22 11:43:29.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-08-22 11:43:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8712e4d14c037c7bcb21bbf0f02e46d44a9b0629","subject":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.","status":"MERGED","created":"2024-08-20 11:34:16.000000000","updated":"2024-08-22 11:43:29.000000000","submitted":"2024-08-22 11:43:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"926622","meta_rev_id":"eae61c7956f9701517509e7eaa8c5836b8166bc2","_number":926622,"virtual_id_number":926622,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-22 11:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-22 08:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-22 11:33:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-08-22 02:48:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-22 11:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-08-22 02:49:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 11:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-20 13:13:30.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-20 17:05:00.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-08-21 14:18:24.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-08-21 14:53:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-08-21 18:38:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2024-08-21 18:52:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-22 08:44:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"643f6c1d1a775ab5f77a475c1d60128a27effe6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-20 11:34:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06d07a7db4bb12b9800e3f35d0c141c8ccaa4db9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 11:35:40.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"25337190666c5004e4696d6e644de4c0ea10cce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 11:42:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f648080aed24a8a84dc03f642f362a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34859098873d4bbaa80d14120dd50086 : SUCCESS in 7m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fdf3410f6fa84d0aa2bd55bba46e1bfc : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/16295b74346c4a3e8fa627c0a58ef087 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"51530ed316e9255db11afd746cf0e316ad5c890c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-20 13:13:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d1bba32f1da2b3d6806e63bd4b80b29ba68b706","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-20 14:05:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68b9712983284d5b54957108c51b051d82acf9db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-21 02:42:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d52339f5417f01cc8fa537ec9ef207e7baca8aee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 02:53:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bf353a802224e9d8f05100043d7d38e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cfe1df83174a329059078da20e5f15 : SUCCESS in 7m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/849cce6221c3407492594e9dd2885c0b : SUCCESS in 6m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3c3912d773d445638fe80a2d9e0087ed : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d7e2875f98514cd0fef6994fb06b2003a5739e3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-21 10:53:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ed8f51a199f5c4ff92da6fb6692bd0c44f3aa132","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 10:54:41.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"972e27f75be07631f7ea8c9e7cbd11ae8a95392a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 11:03:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba8d32de5b94938a3cd79b075244900\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c90e705fb1743b8966a0d2eec76d0e7 : SUCCESS in 7m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d13306a488764a4a8ad1d22490a5e6c7 : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7b10e0068c604fce8850a6014dda7944 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":3},{"id":"7b4747a6352a4152498406ec3b55b2db4d1a62b1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-21 14:18:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8b60754013eb813b84424cdada961db345d7718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:51:01.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_31245\u003e\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"_revision_number":4},{"id":"e7eba71aa4445ca7c52ebb7b31c6b85e5877a595","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:51:49.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"e58b80d8ac32add616e70e5883c7218b64fbafae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 14:53:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6eae82c1bf9f4eb26f60d23a21d23287a3b92512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:59:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e1ccd23ecbe4a33b338f3640fa01d6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7e6ec205d1443da8043b975a96a684 : SUCCESS in 7m 03s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e3267cf0940d4855853410ce7cd023ec : SUCCESS in 7m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fc83460c96be44bab168f6cd7f660e5e : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"04e8af88abbc49b0097c82dd2fa6974f23f7d286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-21 18:38:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"62c643f7d7c552a4793e91ec5fe10fbdf44dc7b4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-21 18:39:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d8a1a941c0ed016f18ce39badd72aa3a17ec9fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 18:48:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0137d50b72d546c78176bd50bedbb7ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85ba60c21614bb4bf5241b4fdcfeb6a : SUCCESS in 7m 04s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/11928f111b71448e944a1443f9ed7a7b : SUCCESS in 5m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/64f78e986ea44aecb590b3c28b6b0dac : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":5},{"id":"d1745bbf9654c977673c45bca314ebf58398f0b1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-21 18:52:12.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"077c94dc952a41e5b2e9b60771d2c8ad6bd88c6e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-21 18:52:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aceee1920c01566dcefaf889c51e5740a1cf1e93","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-22 02:48:41.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ab24fb0e3a4699a28b10712fc413a313d9b0e09e","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 02:49:44.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"da8d536ec7792ae6af57ddafd8df3c12a1f7b682","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-22 08:44:54.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7e104bf1c0a6fb38e176e57f026a2712336f6500","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-22 11:33:58.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c930ebb661c6b59b2b329730e020ba589590b30e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-22 11:34:59.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"810a434ad4dd5b43c61ed1d2a28b1dee20c1ef48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:35:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"690e97fbab850f915524c5482bd2ceb053180ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:43:29.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62259d2e7e084f96993614501e5fa136\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c0e8f848e724d3f8041d6fd18a8aea7 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63753caf81c743279a62cfecda166be2 : SUCCESS in 6m 19s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6912a795e10a42c3bbd69db6ca13be4d : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":5},{"id":"eae61c7956f9701517509e7eaa8c5836b8166bc2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:43:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"acba6e90bfd654080875fa4197be275e7316f66d","revisions":{"5de87b6d378423199ee5a677cebe488d58f46adf":{"kind":"REWORK","_number":1,"created":"2024-08-20 11:34:16.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/22/926622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/926622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/926622/1"}}},"commit":{"parents":[{"commit":"65ed0c311e29276625a239a6df762fad25e8a8db","subject":"Release oslo.config 9.6.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65ed0c311e29276625a239a6df762fad25e8a8db"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 11:34:05.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 11:34:05.000000000","tz":120},"subject":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.","message":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 7.2.0..4e596cd865c0e23f98fab47d09e1cb31ed51b14e§\n\n4e596cd (HEAD -\u003e master, origin/master, origin/HEAD) Add release note for format_inspector\n91af49b Add stream-based detection\na0481d5 Add GPT/MBR inspector\n1290708 Fix qcow2 feature flag checks (for the future)\n1f756b8 Refactor some things for oslo\ncfc0e10 Support VMDK sparse with footer\n23d6b20 Make VMDKInspector support non-sparse formats\nb68be95 Add region-complete signaling for inspectors\ne1f58f4 Reinstate VMDK safety check coverage\ne2bae7a Modularize image safety checks\n8e1b637 Imported Translations from Zanata\ne973df6 Make FileInspector an ABC\n7e1bea6 Test cover target in CI\n15370fb Import format_inspector from nova\nfa53de1 Make imageutils a directory\ne858c7e Fix coverage target\n\nChange-Id: I8712e4d14c037c7bcb21bbf0f02e46d44a9b0629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5de87b6d378423199ee5a677cebe488d58f46adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5de87b6d378423199ee5a677cebe488d58f46adf"}]},"branch":"refs/heads/master"},"bd20643f8b6219331d43d13910d72e184d11a0c8":{"kind":"REWORK","_number":2,"created":"2024-08-21 02:42:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/926622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/926622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/926622/2"}}},"commit":{"parents":[{"commit":"65ed0c311e29276625a239a6df762fad25e8a8db","subject":"Release oslo.config 9.6.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/65ed0c311e29276625a239a6df762fad25e8a8db"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 11:34:05.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-08-21 02:41:42.000000000","tz":540},"subject":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.","message":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 7.2.0..d37a161b7c65a73a6aaabcb743b117cee41a0d58\n\n076aaac Add a function to trim scope out of ipv6 addr\n4e596cd Add release note for format_inspector\n91af49b Add stream-based detection\na0481d5 Add GPT/MBR inspector\n1290708 Fix qcow2 feature flag checks (for the future)\n1f756b8 Refactor some things for oslo\ncfc0e10 Support VMDK sparse with footer\n23d6b20 Make VMDKInspector support non-sparse formats\nb68be95 Add region-complete signaling for inspectors\ne1f58f4 Reinstate VMDK safety check coverage\ne2bae7a Modularize image safety checks\n8e1b637 Imported Translations from Zanata\ne973df6 Make FileInspector an ABC\n7e1bea6 Test cover target in CI\n15370fb Import format_inspector from nova\nfa53de1 Make imageutils a directory\ne858c7e Fix coverage target\n\nChange-Id: I8712e4d14c037c7bcb21bbf0f02e46d44a9b0629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd20643f8b6219331d43d13910d72e184d11a0c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd20643f8b6219331d43d13910d72e184d11a0c8"}]},"branch":"refs/heads/master"},"603500f4d4c1c185cc514163c7b239a569365a0c":{"kind":"REWORK","_number":3,"created":"2024-08-21 10:53:11.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/22/926622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/926622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/926622/3"}}},"commit":{"parents":[{"commit":"238cb9ff9aa4a433773bde22f8cef932ea1d4363","subject":"Release oslo.config 9.6.0 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/238cb9ff9aa4a433773bde22f8cef932ea1d4363"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 11:34:05.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-21 10:52:54.000000000","tz":120},"subject":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.","message":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 7.2.0..4e596cd865c0e23f98fab47d09e1cb31ed51b14e§\n\n4e596cd (HEAD -\u003e master, origin/master, origin/HEAD) Add release note for format_inspector\n91af49b Add stream-based detection\na0481d5 Add GPT/MBR inspector\n1290708 Fix qcow2 feature flag checks (for the future)\n1f756b8 Refactor some things for oslo\ncfc0e10 Support VMDK sparse with footer\n23d6b20 Make VMDKInspector support non-sparse formats\nb68be95 Add region-complete signaling for inspectors\ne1f58f4 Reinstate VMDK safety check coverage\ne2bae7a Modularize image safety checks\n8e1b637 Imported Translations from Zanata\ne973df6 Make FileInspector an ABC\n7e1bea6 Test cover target in CI\n15370fb Import format_inspector from nova\nfa53de1 Make imageutils a directory\ne858c7e Fix coverage target\n\nChange-Id: I8712e4d14c037c7bcb21bbf0f02e46d44a9b0629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/603500f4d4c1c185cc514163c7b239a569365a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/603500f4d4c1c185cc514163c7b239a569365a0c"}]},"branch":"refs/heads/master"},"5d954f93d9be23eb45d237d0e0ccabae2f3d8436":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-21 14:51:01.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/22/926622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/926622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/926622/4"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 11:34:05.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-08-21 14:51:01.000000000","tz":0},"subject":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.","message":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 7.2.0..4e596cd865c0e23f98fab47d09e1cb31ed51b14e§\n\n4e596cd (HEAD -\u003e master, origin/master, origin/HEAD) Add release note for format_inspector\n91af49b Add stream-based detection\na0481d5 Add GPT/MBR inspector\n1290708 Fix qcow2 feature flag checks (for the future)\n1f756b8 Refactor some things for oslo\ncfc0e10 Support VMDK sparse with footer\n23d6b20 Make VMDKInspector support non-sparse formats\nb68be95 Add region-complete signaling for inspectors\ne1f58f4 Reinstate VMDK safety check coverage\ne2bae7a Modularize image safety checks\n8e1b637 Imported Translations from Zanata\ne973df6 Make FileInspector an ABC\n7e1bea6 Test cover target in CI\n15370fb Import format_inspector from nova\nfa53de1 Make imageutils a directory\ne858c7e Fix coverage target\n\nChange-Id: I8712e4d14c037c7bcb21bbf0f02e46d44a9b0629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d954f93d9be23eb45d237d0e0ccabae2f3d8436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d954f93d9be23eb45d237d0e0ccabae2f3d8436"}]},"branch":"refs/heads/master","description":"Rebase"},"acba6e90bfd654080875fa4197be275e7316f66d":{"kind":"REWORK","_number":5,"created":"2024-08-21 18:38:17.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/22/926622/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/926622/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/926622/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/926622/5"}}},"commit":{"parents":[{"commit":"a1a40d6144eb6869b90029579a6c257a963ebd7c","subject":"Release oslo.concurrency 6.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a1a40d6144eb6869b90029579a6c257a963ebd7c"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-20 11:34:05.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-08-21 18:37:30.000000000","tz":120},"subject":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.","message":"Release oslo.utils 7.3.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 7.2.0..d37a161b7c65a73a6aaabcb743b117cee41a0d58\n076aaac Add a function to trim scope out of ipv6 addr\n4e596cd Add release note for format_inspector\n91af49b Add stream-based detection\na0481d5 Add GPT/MBR inspector\n1290708 Fix qcow2 feature flag checks (for the future)\n1f756b8 Refactor some things for oslo\ncfc0e10 Support VMDK sparse with footer\n23d6b20 Make VMDKInspector support non-sparse formats\nb68be95 Add region-complete signaling for inspectors\ne1f58f4 Reinstate VMDK safety check coverage\ne2bae7a Modularize image safety checks\n8e1b637 Imported Translations from Zanata\ne973df6 Make FileInspector an ABC\n7e1bea6 Test cover target in CI\n15370fb Import format_inspector from nova\nfa53de1 Make imageutils a directory\ne858c7e Fix coverage target\n\nChange-Id: I8712e4d14c037c7bcb21bbf0f02e46d44a9b0629\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/acba6e90bfd654080875fa4197be275e7316f66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/acba6e90bfd654080875fa4197be275e7316f66d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
