)]}'
{"id":"openstack%2Freleases~926765","triplet_id":"openstack%2Freleases~master~I39894cb4866c479dba5c19eb4ae9fe1a41cb975c","project":"openstack/releases","branch":"master","topic":"dalmatian-final-non-client-libs","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-09-02 16:40:03.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-09-02 16:33:06.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-02 16:40:03.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-08-30 14:34:45.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I39894cb4866c479dba5c19eb4ae9fe1a41cb975c","subject":"Release final openstacksdk for 2024.2 (Dalmatian)","status":"MERGED","created":"2024-08-21 13:18:02.000000000","updated":"2024-09-02 16:40:03.000000000","submitted":"2024-09-02 16:40:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926765-dalmatian-final-non-client-libs","meta_rev_id":"92d6e06bc4550f06e95f5f1ff872646037ec88c8","_number":926765,"virtual_id_number":926765,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-02 16:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-09-02 16:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-30 13:13:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-08-30 13:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-30 14:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-09-02 16:33:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-08-30 14:35:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 13:18:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-21 13:20:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-08-21 13:44:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-23 14:04:02.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2024-08-26 09:33:06.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-08-30 13:13:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-30 13:18:10.000000000","updated_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"reviewer":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"state":"REVIEWER"}],"messages":[{"id":"a485ead233ba6723b16c63cb353e10cba68bf7be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-21 13:18:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f41e312a707b8e1307aead976964b2da6a68be9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 13:44:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85a26be9ba674b188a10ad5f66c28197\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb030968b7944139743474d66432635 : SUCCESS in 6m 47s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b2f0f76421f8489191f73fd92582c373 : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f8671aa5bbeb4b889e27c1756ce42e01 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"84bd052295c5ed5d5339c73373f232e675c24ec9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-23 06:30:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a004e154213b92e1a3823241d4a8e7657260d7b6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 06:31:37.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"eda287fd945729c1e2ca5b8f66d6245dc7590a83","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-23 11:31:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b254f0e0186c646a68279c81e7e1083e048448b3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-08-23 14:04:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e2aba3907cfb4b3620e3b3a4320e9b675ae74b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-26 09:32:35.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d72a93ad5fe8ab6dc95de2e1f80bbead717b46d3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-26 09:33:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2ed68c8698b5b157538b89110dd90c22012fd81","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-26 09:40:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"74244c358fe8773a2952df3fc0f08a72028546d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 10:17:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de4ec3e3b58a4a978874fa906e60efde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7bd407dde444f1bb268a239d281c20 : SUCCESS in 7m 12s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b86aa98156bb412599da06fddf16deed : SUCCESS in 4m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ee2edeff6e0d4183abecc221b8da7000 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"f53c730317ed49247e886e3699ee26868f4c1016","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-30 13:13:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bcda475aa376a11f2e01a57d8604d3340b6b9530","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2024-08-30 13:18:10.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6435e33bec214172be5c1d181b110b8539b7794","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-08-30 14:34:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f5d658c7d4eb175ae017fc36ad047e890b263c3d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:35:32.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"73f004421453ab9ba757b61376578187c41d2432","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-02 16:33:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ebea5eebd3f7bc61dbac02584e91fb3ac8c8fd29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 16:33:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d64b652597ca5b5732cc93b25327d1fbe0fe0597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 16:40:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e1469d9ee34b3ab961c9e2aac06084\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/918c8358ba2942e68c0d7e28037545cb : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b0d9c5197142a5b581be0cd5c9a720 : SUCCESS in 6m 13s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ffe21003c83b49dcac4d2390fe473396 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"92d6e06bc4550f06e95f5f1ff872646037ec88c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 16:40:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"256021084794ea9854c1b5ab02e5943069c8ef43","revisions":{"8d041277be258ff4a16de4e7e9ecd9490f6287ca":{"kind":"REWORK","_number":1,"created":"2024-08-21 13:18:02.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/65/926765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/926765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/926765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/926765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/926765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/926765/1"}}},"commit":{"parents":[{"commit":"c126da6b788367725c4b793084e3f652292f673b","subject":"Merge \"Release os-brick for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c126da6b788367725c4b793084e3f652292f673b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-21 13:17:55.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-21 13:17:55.000000000","tz":120},"subject":"Release final openstacksdk for 2024.2 (Dalmatian)","message":"Release final openstacksdk for 2024.2 (Dalmatian)\n\nThis proposes a release for openstacksdk to get any final merged changes\nout prior to the lib freeze for 2024.2 going into effect.\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\n$ git log --oneline --no-merges 3.3.0..b65b7d4d3\ne11bb7c3c Deprecate \u0027use_direct_get\u0027 parameter\n05c442ddf Support server unshelve to specific host\nbbc130ed6 Drop support for Python 3.7\nb44df1479 Add Python 3.12 classifier\n12ed1a343 image: Check path is a file before attempting to use it (redux)\nfc18b74c9 mypy: Enable checks for openstack.tests.functional\nc4b9a44a1 tests: Remove TestClustering functional tests\n5704b3ef7 mypy: Enable checks for openstack.cloud\n71b31f15e cloud: Combine networking-related classes\naee822095 cloud: Misc type fixes\n2f1829c1d cloud: Reorganize _OpenStackCloudMixin, Connection (2/2)\nd6e11e487 cloud: Reorganize _OpenStackCloudMixin, Connection (1/2)\n53a2ea6f6 Added missing stack documentation and rewrite stack proxy document\n4a7e36f31 cloud: Make service mixins subclass _OpenStackCloudMixin\nc84215409 cloud: Make _OpenStackCloudMixin subclass ServicesMixin\n200e1524a mypy: Enable checks for openstack.tests.fixtures\n04a1cd2fb cloud: Fix annotations for proxy method\nedd72883d cloud: Remove check for nova extensions\n1d9596288 pre-commit: Bump versions\nc65b4d7c2 cloud: Replace reference to removed variable\nf41d60c7e block storage: Add missing \u0027find_type\u0027 proxy API to v2\nff1ac4311 block storage: Add \u0027set_readonly\u0027 action to v2\n346756097 baremetal: Enhance VIF attachment with port and portgroup UUIDs\necb35e0a6 block_storage: Add support for project_id in Limits\n23c7a3aaa zuul: Use more stable dib job\n4a8c12a80 tests: Remove errant print\n8ccf5132e cloud: Fix typo\n9a023c503 image: Check path is a file before attempting to use it\n14c1ebbeb exceptions: ResourceNotFound -\u003e NotFoundException\n\nChange-Id: I39894cb4866c479dba5c19eb4ae9fe1a41cb975c\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d041277be258ff4a16de4e7e9ecd9490f6287ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8d041277be258ff4a16de4e7e9ecd9490f6287ca"}]},"branch":"refs/heads/master"},"256021084794ea9854c1b5ab02e5943069c8ef43":{"kind":"REWORK","_number":2,"created":"2024-08-26 09:32:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/65/926765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/926765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/926765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/926765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/926765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/926765/2"}}},"commit":{"parents":[{"commit":"c126da6b788367725c4b793084e3f652292f673b","subject":"Merge \"Release os-brick for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c126da6b788367725c4b793084e3f652292f673b"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-08-21 13:17:55.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-26 09:32:35.000000000","tz":0},"subject":"Release final openstacksdk for 2024.2 (Dalmatian)","message":"Release final openstacksdk for 2024.2 (Dalmatian)\n\nThis proposes a release for openstacksdk to get any final merged changes\nout prior to the lib freeze for 2024.2 going into effect.\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\n$ git log --oneline --no-merges 3.3.0..b65b7d4d3\ne11bb7c3c Deprecate \u0027use_direct_get\u0027 parameter\n05c442ddf Support server unshelve to specific host\nbbc130ed6 Drop support for Python 3.7\nb44df1479 Add Python 3.12 classifier\n12ed1a343 image: Check path is a file before attempting to use it (redux)\nfc18b74c9 mypy: Enable checks for openstack.tests.functional\nc4b9a44a1 tests: Remove TestClustering functional tests\n5704b3ef7 mypy: Enable checks for openstack.cloud\n71b31f15e cloud: Combine networking-related classes\naee822095 cloud: Misc type fixes\n2f1829c1d cloud: Reorganize _OpenStackCloudMixin, Connection (2/2)\nd6e11e487 cloud: Reorganize _OpenStackCloudMixin, Connection (1/2)\n53a2ea6f6 Added missing stack documentation and rewrite stack proxy document\n4a7e36f31 cloud: Make service mixins subclass _OpenStackCloudMixin\nc84215409 cloud: Make _OpenStackCloudMixin subclass ServicesMixin\n200e1524a mypy: Enable checks for openstack.tests.fixtures\n04a1cd2fb cloud: Fix annotations for proxy method\nedd72883d cloud: Remove check for nova extensions\n1d9596288 pre-commit: Bump versions\nc65b4d7c2 cloud: Replace reference to removed variable\nf41d60c7e block storage: Add missing \u0027find_type\u0027 proxy API to v2\nff1ac4311 block storage: Add \u0027set_readonly\u0027 action to v2\n346756097 baremetal: Enhance VIF attachment with port and portgroup UUIDs\necb35e0a6 block_storage: Add support for project_id in Limits\n23c7a3aaa zuul: Use more stable dib job\n4a8c12a80 tests: Remove errant print\n8ccf5132e cloud: Fix typo\n9a023c503 image: Check path is a file before attempting to use it\n14c1ebbeb exceptions: ResourceNotFound -\u003e NotFoundException\n\nChange-Id: I39894cb4866c479dba5c19eb4ae9fe1a41cb975c\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/256021084794ea9854c1b5ab02e5943069c8ef43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/256021084794ea9854c1b5ab02e5943069c8ef43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
