)]}'
{"id":"openstack%2Freleases~927136","triplet_id":"openstack%2Freleases~master~Ie4a1fe2fd328a1399bc31b327eed2d074aa6271e","project":"openstack/releases","branch":"master","topic":"dalmatian-milestone-3","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-08-30 14:55:38.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-08-30 14:55:38.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2024-08-30 13:02:11.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2024-08-30 14:55:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie4a1fe2fd328a1399bc31b327eed2d074aa6271e","subject":"Release final python-novaclient for 2024.2 Dalmatian","status":"MERGED","created":"2024-08-26 13:38:03.000000000","updated":"2024-08-30 14:55:38.000000000","submitted":"2024-08-30 14:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927136-dalmatian-milestone-3","meta_rev_id":"b0d85aedfa0a1513d88bd308f6e37367196e7505","_number":927136,"virtual_id_number":927136,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 14:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-30 14:36:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-30 12:30:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-08-30 12:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-30 13:02:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-30 14:36:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-08-30 12:28:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 13:38:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-26 13:56:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 14:20:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-08-26 14:20:42.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2024-08-26 14:20:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2024-08-30 14:36:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"fa171ba73052f926059975ac0117d0bfe25061b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-26 13:38:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac40cf2f20c3d2ad579c03807c197e2827706bf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 13:56:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9830ec6cf78d4f7786a6560679713b20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c211f78fe01944b0af75680445a8feb1 : SUCCESS in 6m 42s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/25d5dc34faa548c29be7ddad05e3038d : SUCCESS in 5m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9beb22738fbc40428f6a578630a55a81 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"3a0a61d896de1043fa391a79901a1e1c59f194a2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-27 16:13:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2568b03e3c774bcf031ad1e237a5efb2f1c1fa85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-30 11:35:49.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"48f3febd996a0ed564177b808ba3d49a741e28ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 11:44:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa92e45deb284441b7fdfef310860412\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4689e89ccfe541be88f2110574edf425 : SUCCESS in 6m 59s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/502ee71c02934796afa72d17bbef1135 : SUCCESS in 6m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1c588d15de564db889bb7498a18a8692 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"d29cb7821dae303731c8d53d39ce2c28c6b50ce6","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-30 12:14:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9f22430ed5a4c4a063f49893ecacd4013fbfcca","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-08-30 12:28:07.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"616dcb5d2e036e4176563c0a29f5ee4832a2971a","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 12:28:54.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"d472af665a8b19854172de7ca2737ed68262ff33","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-30 12:30:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"de061859fd0ee508689669f4c477dfcbccfccee4","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-08-30 13:02:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"338f4e51a5ddce9b02ac170a10fc96368c4a290c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-30 14:36:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3d9ff0d8b54c9d52f217cc12e87d762fa00468cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:37:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"912bab1f727f1ee643a527baa59879f0d26efbd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:55:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/711b32e2e00549a29ec28cfe284a2468\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d677a7c3f9e47d585eef92d234f15b3 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c01273b9e9480897af0f42ba753639 : SUCCESS in 7m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1d94250d753944cabb31e24980ce582a : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b0d85aedfa0a1513d88bd308f6e37367196e7505","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 14:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5681a38c03eea35d927ba45b4556a13222dd02cb","revisions":{"9f373718ca212f270ed0ab37266f53c701cca56d":{"kind":"REWORK","_number":1,"created":"2024-08-26 13:38:03.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/36/927136/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/927136/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/927136/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/927136/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/927136/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/927136/1"}}},"commit":{"parents":[{"commit":"7ef5f04bbeefc025e523b8adf128af020b9b4024","subject":"Merge \"Release final os-vif for 2024.2 (Dalmatian)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ef5f04bbeefc025e523b8adf128af020b9b4024"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-26 13:37:59.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-26 13:37:59.000000000","tz":120},"subject":"Release final python-novaclient for 2024.2 Dalmatian","message":"Release final python-novaclient for 2024.2 Dalmatian\n\nThis is a release for python-novaclient to get any merged but\nunreleased changes out prior to client library freeze for\n2024.2 Dalmatian going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged after Thursday, August 29th.\n\n$ git log --oneline --no-merges 18.6.0..2bd135c1\n2bd135c1 Remove old excludes\nec292ac6 reno: Update master for unmaintained/zed\n3ffc5e5b Update master for stable/2024.1\n99bbecfc reno: Update master for unmaintained/xena\nf2d51a0f reno: Update master for unmaintained/wallaby\n59a147ab reno: Update master for unmaintained/victoria\n\nChange-Id: Ie4a1fe2fd328a1399bc31b327eed2d074aa6271e\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f373718ca212f270ed0ab37266f53c701cca56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f373718ca212f270ed0ab37266f53c701cca56d"}]},"branch":"refs/heads/master"},"5681a38c03eea35d927ba45b4556a13222dd02cb":{"kind":"REWORK","_number":2,"created":"2024-08-30 11:35:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/36/927136/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/927136/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/927136/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/927136/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/927136/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/927136/2"}}},"commit":{"parents":[{"commit":"7ef5f04bbeefc025e523b8adf128af020b9b4024","subject":"Merge \"Release final os-vif for 2024.2 (Dalmatian)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ef5f04bbeefc025e523b8adf128af020b9b4024"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-26 13:37:59.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-08-30 11:35:49.000000000","tz":0},"subject":"Release final python-novaclient for 2024.2 Dalmatian","message":"Release final python-novaclient for 2024.2 Dalmatian\n\nThis is a release for python-novaclient to get any merged but\nunreleased changes out prior to client library freeze for\n2024.2 Dalmatian going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged after Thursday, August 29th.\n\n$ git log --oneline --no-merges 18.6.0..2bd135c1\n2bd135c1 Remove old excludes\nec292ac6 reno: Update master for unmaintained/zed\n3ffc5e5b Update master for stable/2024.1\n99bbecfc reno: Update master for unmaintained/xena\nf2d51a0f reno: Update master for unmaintained/wallaby\n59a147ab reno: Update master for unmaintained/victoria\n\nChange-Id: Ie4a1fe2fd328a1399bc31b327eed2d074aa6271e\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5681a38c03eea35d927ba45b4556a13222dd02cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5681a38c03eea35d927ba45b4556a13222dd02cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
