)]}'
{"id":"openstack%2Freleases~927166","triplet_id":"openstack%2Freleases~master~I6a10fe1ecfc06297325a187383f104e21f18e2be","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-29 15:26:26.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-08-29 15:15:27.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2024-08-29 15:26:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a10fe1ecfc06297325a187383f104e21f18e2be","subject":"Release microversion-parse 2.0.0 for 2024.2 dalmatian.","status":"MERGED","created":"2024-08-26 15:29:36.000000000","updated":"2024-08-29 15:26:26.000000000","submitted":"2024-08-29 15:26:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927166","meta_rev_id":"f949979d413687abb31b6e3cd442d6757d19bd1e","_number":927166,"virtual_id_number":927166,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-29 15:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-08-27 14:40:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-08-29 15:15:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2024-08-29 15:15:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-08-27 14:41:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 15:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 16:20:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-08-27 09:52:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2024-08-27 13:02:17.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-08-29 15:15:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"788dbf59ab6a15486d8c6f54703ed409ad7a23fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-26 15:29:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1d3b63a7bdd3e244ed98dd1abc16fd34775f794","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 15:31:01.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"2b8c4de38594ed0b9731c32b52250990081a12a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 17:24:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a2fa1b5624d46a28d2e64c6557f52af\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/366ecbe2f9184507ac8275030e6b6309 : SUCCESS in 6m 28s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/136c55ca7fe043329fe00e6803aa8022 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/16bdf9b56e4c436e98efb042b1296f15 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"671c60fcd0c64b8487a1a62045db9b38c1586143","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-27 09:13:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25088252fe1d82363a10b6efb6fbb8b65f72360a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-27 09:52:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22ae9aee55e21f66ce6cff4045c1413fa237d8c8","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-27 11:00:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b242ba3b4ef9b81b6b73ef1d508e0583374effe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-27 12:03:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2370b9756ef960ba6ebaeb14b59a550820532fe9","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-27 12:03:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d16d9f4a03af809b21d04a340564d182a7ee500","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-27 13:02:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0725e61d804ad33c5b489031b695af5efc0a6e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-27 14:40:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_31245\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"_revision_number":3},{"id":"73aa45e5c6b2137f58799289fb2549b8494f01f6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 14:41:04.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"766706a775b777f2cece918c9bc3f546d3dbfb9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 14:52:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d7962d023b84aa292b58d691362318b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58014d04b2cd4b0da1cdaa9594ee6445 : SUCCESS in 6m 24s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d97948d6160b46fa88d9d81cbd59ccf8 : SUCCESS in 5m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/452c9c3950224dda8d1ed9a9666301c3 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"78178a116fe25a9f6d15687c0a4aabc795cee9dd","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-08-29 08:20:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c691902d86fb5ee7dfc0e76465ce15cf90b9eb8f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-08-29 15:15:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"db898d2c5d9dde4229d6f5a0e66c63fdf5a27d17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 15:15:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9f1f6d24dd4e9452eae6994e219bca2381d336eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 15:26:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00efce69b68d49ce9a8d320d0c4e6656\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/941a68d425c34f778d7497070d6fc969 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9465c89898fb49f7a42c2050b201edad : SUCCESS in 9m 12s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/633375f7b83a45f5883ce20c4441098d : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"f949979d413687abb31b6e3cd442d6757d19bd1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 15:26:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b07acba7ff2e23a73342715a7b07b70b78ed6089","revisions":{"88d52736b06b22cf5abb672f8a7cf8f0da306019":{"kind":"REWORK","_number":1,"created":"2024-08-26 15:29:36.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/66/927166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/927166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/927166/1"}}},"commit":{"parents":[{"commit":"fa5369e5f33f027d25e3e74a6b35366beccf8e3b","subject":"Release futurist 3.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa5369e5f33f027d25e3e74a6b35366beccf8e3b"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-26 15:27:02.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-26 15:29:10.000000000","tz":120},"subject":"Release microversion-parse 1.1.0 for 2024.2 dalmatian.","message":"Release microversion-parse 1.1.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 1.0.1..26a2dc96cc2a0e6baf528c846f2d6d48e1dd3379\n\n* 26a2dc9 2024-01-28 15:54:38 +0900 Bump hacking\n* 05107bf 2024-01-10 22:47:44 +0900 Update python classifier in setup.cfg\n* 3d3edb7 2023-08-07 15:38:15 +0800 setup.cfg: Replace dashes with underscores\n| * 943a2f4 2023-09-07 19:48:13 +0900 Fix warnings in doc build\n|/\n* edddb6b 2023-05-04 20:21:09 +0000 Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n* 18f5be7 2023-04-10 13:17:40 +0900 Moves supported python runtimes from version 3.8 to 3.10\n* d8e1232 2022-07-28 08:58:56 +0800 remove unicode from code\n* 6d98355 2022-05-09 15:10:02 +0200 Drop python3.6/3.7 support in testing runtime\n* dc07afb 2021-01-21 15:42:04 +0100 Use py3 as the default runtime for tox\n| * ee08574 2021-03-23 13:13:42 +0100 Move flake8 as a pre-commit local target.\n|/\n* 4ea5fb3 2020-11-03 14:32:48 +0100 Use Python 3.8 as the default runtime for tox\n* 89cb11c 2020-09-15 15:19:35 +0200 Adding pre-commit\n\nChange-Id: I6a10fe1ecfc06297325a187383f104e21f18e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88d52736b06b22cf5abb672f8a7cf8f0da306019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/88d52736b06b22cf5abb672f8a7cf8f0da306019"}]},"branch":"refs/heads/master"},"849864fc6b5eb64190703844ed89ec2523164f87":{"kind":"REWORK","_number":2,"created":"2024-08-27 12:03:21.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/66/927166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/927166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/927166/2"}}},"commit":{"parents":[{"commit":"fa5369e5f33f027d25e3e74a6b35366beccf8e3b","subject":"Release futurist 3.0.1 for 2024.2 dalmatian.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fa5369e5f33f027d25e3e74a6b35366beccf8e3b"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-26 15:27:02.000000000","tz":120},"committer":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-27 12:03:06.000000000","tz":120},"subject":"Release microversion-parse 2.0.0 for 2024.2 dalmatian.","message":"Release microversion-parse 2.0.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 1.0.1..26a2dc96cc2a0e6baf528c846f2d6d48e1dd3379\n\n* 26a2dc9 2024-01-28 15:54:38 +0900 Bump hacking\n* 05107bf 2024-01-10 22:47:44 +0900 Update python classifier in setup.cfg\n* 3d3edb7 2023-08-07 15:38:15 +0800 setup.cfg: Replace dashes with underscores\n| * 943a2f4 2023-09-07 19:48:13 +0900 Fix warnings in doc build\n|/\n* edddb6b 2023-05-04 20:21:09 +0000 Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n* 18f5be7 2023-04-10 13:17:40 +0900 Moves supported python runtimes from version 3.8 to 3.10\n* d8e1232 2022-07-28 08:58:56 +0800 remove unicode from code\n* 6d98355 2022-05-09 15:10:02 +0200 Drop python3.6/3.7 support in testing runtime\n* dc07afb 2021-01-21 15:42:04 +0100 Use py3 as the default runtime for tox\n| * ee08574 2021-03-23 13:13:42 +0100 Move flake8 as a pre-commit local target.\n|/\n* 4ea5fb3 2020-11-03 14:32:48 +0100 Use Python 3.8 as the default runtime for tox\n* 89cb11c 2020-09-15 15:19:35 +0200 Adding pre-commit\n\nChange-Id: I6a10fe1ecfc06297325a187383f104e21f18e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/849864fc6b5eb64190703844ed89ec2523164f87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/849864fc6b5eb64190703844ed89ec2523164f87"}]},"branch":"refs/heads/master"},"b07acba7ff2e23a73342715a7b07b70b78ed6089":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-27 14:40:18.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/66/927166/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/927166/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/927166/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/927166/3"}}},"commit":{"parents":[{"commit":"b591c43a0682b06dbb8d21792ce750ff9ad0ea4b","subject":"Merge \"Release pbr 2.5.0 for 2024.2 dalmatian.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b591c43a0682b06dbb8d21792ce750ff9ad0ea4b"}]}],"author":{"name":"damani42","email":"dbengt@redhat.com","date":"2024-08-26 15:27:02.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2024-08-27 14:40:18.000000000","tz":0},"subject":"Release microversion-parse 2.0.0 for 2024.2 dalmatian.","message":"Release microversion-parse 2.0.0 for 2024.2 dalmatian.\n\n$ git log --oneline --no-merges 1.0.1..26a2dc96cc2a0e6baf528c846f2d6d48e1dd3379\n\n* 26a2dc9 2024-01-28 15:54:38 +0900 Bump hacking\n* 05107bf 2024-01-10 22:47:44 +0900 Update python classifier in setup.cfg\n* 3d3edb7 2023-08-07 15:38:15 +0800 setup.cfg: Replace dashes with underscores\n| * 943a2f4 2023-09-07 19:48:13 +0900 Fix warnings in doc build\n|/\n* edddb6b 2023-05-04 20:21:09 +0000 Revert \"Moves supported python runtimes from version 3.8 to 3.10\"\n* 18f5be7 2023-04-10 13:17:40 +0900 Moves supported python runtimes from version 3.8 to 3.10\n* d8e1232 2022-07-28 08:58:56 +0800 remove unicode from code\n* 6d98355 2022-05-09 15:10:02 +0200 Drop python3.6/3.7 support in testing runtime\n* dc07afb 2021-01-21 15:42:04 +0100 Use py3 as the default runtime for tox\n| * ee08574 2021-03-23 13:13:42 +0100 Move flake8 as a pre-commit local target.\n|/\n* 4ea5fb3 2020-11-03 14:32:48 +0100 Use Python 3.8 as the default runtime for tox\n* 89cb11c 2020-09-15 15:19:35 +0200 Adding pre-commit\n\nChange-Id: I6a10fe1ecfc06297325a187383f104e21f18e2be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b07acba7ff2e23a73342715a7b07b70b78ed6089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b07acba7ff2e23a73342715a7b07b70b78ed6089"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
