)]}'
{"id":"openstack%2Freleases~928575","triplet_id":"openstack%2Freleases~master~If9627feab9099fa2eefa26ff4e59370ffbfe7e46","project":"openstack/releases","branch":"master","topic":"dalmatian-rc1-deadline","attention_set":{},"removed_from_attention_set":{"11536":{"account":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"last_update":"2024-09-25 09:42:12.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-09-25 09:42:12.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-09-25 09:42:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If9627feab9099fa2eefa26ff4e59370ffbfe7e46","subject":"Release zun RC1 for 2024.2 Dalmatian","status":"MERGED","created":"2024-09-09 08:51:49.000000000","updated":"2024-09-25 09:42:12.000000000","submitted":"2024-09-25 09:42:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928575-dalmatian-rc1-deadline","meta_rev_id":"139d1e2ebcacf84167a9d5065b4e01f1fff9c3bd","_number":928575,"virtual_id_number":928575,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-25 09:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-09-25 09:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-09-25 09:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-09-25 09:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-09-25 09:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 08:52:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-09 08:58:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-09 11:20:22.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"state":"REVIEWER"},{"updated":"2024-09-25 07:30:47.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"c6b2ddd021e8f3c5096ab7434acf76c4ba1bf6a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-09 08:51:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c47dfa0a759e823e2ac64870d0e7b8b2ff7b08e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 08:58:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a353fe8358644e6aa71f1360ef7197d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb2a1ddf5e414edf86e5f07364cf9a16 : SUCCESS in 6m 22s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/8b1936c572524936b62711e6ec029299 : SUCCESS in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/525e325819454a419ebaf4b42c04681f : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d91ce8c890b1b6b38bc470b54288b20ddd56653f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-10 08:19:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f66db9ca0d3beff579b4c40b72d10d09c52de6b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-09-25 07:30:47.000000000","message":"Patch Set 1: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65273789dc1a9fb598bc4a5d80c733a3fa9d352b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-09-25 07:44:42.000000000","message":"Patch Set 1: -Code-Review -PTL-Approved\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c2a7f35241791755b9da9f3c1e1a5de25f94852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-25 09:11:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"98d94d3bf877bd72df467c007b10d9ff0ba37e83","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-09-25 09:14:32.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eeb9209d798513d4f14dc40ea293c12b677a6fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 09:19:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cc7bc21a85a45babd1a7af068335df9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97705c8a191946ec99a723a154664360 : SUCCESS in 6m 15s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/187f3e3ddc99423985f1e3f863e65d6c : SUCCESS in 4m 31s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/940b9918a70f4e02a941973fbd3c5d89 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"9ba6090fce347ce1b9fbbc81a28053950d4938dd","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-09-25 09:32:16.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb650743f3dd5b5a0736b5d007d01d48e5a6f7cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 09:32:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e2f7bb8948ee17bfe85004724c31784591f7080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 09:42:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ed3b8d73b0144748a372851012cecd9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ab17f47b1e4591a9096f826e48cac9 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c313dc7c6b04e8a96bda8960ffef362 : SUCCESS in 6m 42s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/03d89add9a384b548b31415fd449c08c : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"139d1e2ebcacf84167a9d5065b4e01f1fff9c3bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-25 09:42:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79ca51ee4fd4f34470818dda97031727366d27b5","revisions":{"d11ae6d6341920e46c42677106bc7bace681243d":{"kind":"REWORK","_number":1,"created":"2024-09-09 08:51:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/75/928575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/928575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/928575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/928575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/928575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/928575/1"}}},"commit":{"parents":[{"commit":"22f88e33cb870321147bd1ff84f686103a5ab873","subject":"Merge \"Release Tempest 40.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22f88e33cb870321147bd1ff84f686103a5ab873"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 08:51:45.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 08:51:45.000000000","tz":120},"subject":"Release zun RC1 for 2024.2 Dalmatian","message":"Release zun RC1 for 2024.2 Dalmatian\n\nThis proposed a RC1 release for zun to get any merged changes\nout prior to the RC1 deadline for 2024.2 Dalmatian going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, September 13th.\n\nChange-Id: If9627feab9099fa2eefa26ff4e59370ffbfe7e46\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d11ae6d6341920e46c42677106bc7bace681243d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d11ae6d6341920e46c42677106bc7bace681243d"}]},"branch":"refs/heads/master"},"79ca51ee4fd4f34470818dda97031727366d27b5":{"kind":"REWORK","_number":2,"created":"2024-09-25 09:11:21.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/75/928575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/928575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/928575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/928575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/928575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/928575/2"}}},"commit":{"parents":[{"commit":"22f88e33cb870321147bd1ff84f686103a5ab873","subject":"Merge \"Release Tempest 40.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/22f88e33cb870321147bd1ff84f686103a5ab873"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-09-09 08:51:45.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-09-25 09:10:59.000000000","tz":120},"subject":"Release zun RC1 for 2024.2 Dalmatian","message":"Release zun RC1 for 2024.2 Dalmatian\n\nThis proposed a RC1 release for zun to get any merged changes\nout prior to the RC1 deadline for 2024.2 Dalmatian going into effect.\n\nIf the team is ready to proceed, please leave a +1 to acknowledge this\nrelease and we will process right away. If the team is working on any\nfinal merges, leave a -1 with some detail on the status and update the\npatch with the new commit hash to use when ready.\n\nPatches with no response will be merged by Friday, September 13th.\n\nChange-Id: If9627feab9099fa2eefa26ff4e59370ffbfe7e46\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79ca51ee4fd4f34470818dda97031727366d27b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/79ca51ee4fd4f34470818dda97031727366d27b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
