)]}'
{"id":"openstack%2Freleases~929915","triplet_id":"openstack%2Freleases~master~I42415e4682216da683e83eb3288c2fd409c5daf5","project":"openstack/releases","branch":"master","topic":"upper-constraints-considered-harmful","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-09-19 14:46:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I42415e4682216da683e83eb3288c2fd409c5daf5","subject":"Add mapping of team name to IRC","status":"ABANDONED","created":"2024-09-19 12:45:32.000000000","updated":"2024-09-19 14:49:09.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"175bba08b67947a7a0aab305bd62b8516cefd2c2","_number":929915,"virtual_id_number":929915,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-09-19 13:55:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-19 12:58:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-19 13:59:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"}],"messages":[{"id":"6688a2ee5b1007ee915877836d90e0dd4849604e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-19 12:45:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b940ede66efb38b677dcc062e870d2399cba378e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 12:58:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bae8b65c32f04d9f952b6f3fdb933482\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/472475bbfa2b4c7a827f5e0741ab45fc : FAILURE in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77a6661e46df4bb7815d520e6e8715e4 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bd01b0a302340f2a20309cfb84d47d4 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f76551e40c39463db0218018ec6230da : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f404eb62050646669069d60ffd94e5b5 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b888fc848524ba4a5ea2249136b8fa7 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5bb51060c54d5bb589ba7414c3ed68 : SUCCESS in 4m 57s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/819af773a4f34bdeae030edb2e94a245 : SUCCESS in 4m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a78697a707ae4eaa9d1c88d178533904 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"631ee2acdc7b6c31444ae9275dfbe4afa83a026a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-19 13:44:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"51421791f1bc88a09de0a0bed019e902d657bd80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:55:12.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e15b26ecae6a455989ad8b7f31984a41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79adf9de6ff64c0480532616048b0a24 : FAILURE in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2ab3a316abd4f84a1af1d572e540318 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94e22ae582c44b18bc16e249bb47ca10 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fa1c8f4f5af4100afa63f97e465b442 : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/482802f5014d44d0b472b87ad8493ced : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/069aa14276a540be94d73a6578a2a6ef : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a074489c4d747ba92de5da43e183977 : SUCCESS in 6m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/dd91a6b428fe47faad269d70ee66ebcc : SUCCESS in 4m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/70302b454df14baa9b63a08b8de5eb11 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"21c65f9185fccc3ee5e956cc9aa6950e9b715e03","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-09-19 13:59:49.000000000","message":"Patch Set 2:\n\nThis is also sort of duplicating what\u0027s already in https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml which might make more sense to key off?","accounts_in_message":[],"_revision_number":2},{"id":"8738bcef21c6710c9d11d16582d3e11fef74bc17","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-19 14:46:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c47578a3bc77d9f2db8a107cf25ee8afd64527c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-19 14:47:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"175bba08b67947a7a0aab305bd62b8516cefd2c2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-19 14:49:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"21b197ce977bf8763d2ea9f652dcf6e3cbcc433d","revisions":{"148541583b6adae937fc3a696c7936616f08c5e9":{"kind":"REWORK","_number":1,"created":"2024-09-19 12:45:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/929915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/929915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/929915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/929915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/929915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/929915/1"}}},"commit":{"parents":[{"commit":"727e4436c9f31106eb77006939efc52623d0685e","subject":"Add release note links for 2024.2 Dalmatian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/727e4436c9f31106eb77006939efc52623d0685e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-19 11:18:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-19 12:43:28.000000000","tz":60},"subject":"Add mapping of team name to IRC","message":"Add mapping of team name to IRC\n\nThe inverse of this already sort of exists in the gerritbot\nconfiguration in project-config, but we need it the other way and for\nslightly different purposes.\n\nThis builds upon work done in project-config to associate teams with\nreleases [1]. Now that we have a mapping of teams to IRC channels, we\ncan start reporting failures directly to those teams.\n\n[1] https://review.opendev.org/c/openstack/project-config/+/929914\n\nChange-Id: I42415e4682216da683e83eb3288c2fd409c5daf5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/148541583b6adae937fc3a696c7936616f08c5e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/148541583b6adae937fc3a696c7936616f08c5e9"}]},"branch":"refs/heads/master"},"21b197ce977bf8763d2ea9f652dcf6e3cbcc433d":{"kind":"REWORK","_number":2,"created":"2024-09-19 13:44:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/929915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/15/929915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/929915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/929915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/15/929915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/15/929915/2"}}},"commit":{"parents":[{"commit":"727e4436c9f31106eb77006939efc52623d0685e","subject":"Add release note links for 2024.2 Dalmatian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/727e4436c9f31106eb77006939efc52623d0685e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-19 11:18:38.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-19 13:43:33.000000000","tz":60},"subject":"Add mapping of team name to IRC","message":"Add mapping of team name to IRC\n\nThe inverse of this already sort of exists in the gerritbot\nconfiguration in project-config, but we need it the other way and for\nslightly different purposes.\n\nThis builds upon work done in project-config to associate teams with\nreleases [1]. Now that we have a mapping of teams to IRC channels, we\ncan start reporting failures directly to those teams.\n\n[1] https://review.opendev.org/c/openstack/project-config/+/929914\n\nChange-Id: I42415e4682216da683e83eb3288c2fd409c5daf5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21b197ce977bf8763d2ea9f652dcf6e3cbcc433d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/21b197ce977bf8763d2ea9f652dcf6e3cbcc433d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
