)]}'
{"id":"openstack%2Freleases~931066","triplet_id":"openstack%2Freleases~master~Ic28c8ad4bc84f094a37b38037a6859d725e85f51","project":"openstack/releases","branch":"master","topic":"slurp-graph","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-10-02 15:56:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic28c8ad4bc84f094a37b38037a6859d725e85f51","subject":"Render a graph showing the SLURP and not SLURP upgrade paths","status":"MERGED","created":"2024-10-01 21:32:24.000000000","updated":"2024-10-02 15:56:52.000000000","submitted":"2024-10-02 15:56:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931066-slurp-graph","meta_rev_id":"e1bb5e8ec3d6f841d14ce4f7797caadf43766f4b","_number":931066,"virtual_id_number":931066,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-02 15:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-10-02 14:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-10-02 15:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-10-02 15:18:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-10-02 14:53:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-01 21:38:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 14:53:46.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-10-02 15:18:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"b91b3c6e7858ed6a003ad013c0a22150ec450bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-10-01 21:32:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d002338db856dbdfd86eab47abf37f937ec85cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 21:38:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d2bbbe833d042f78f026cea87d51f0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3c7d1e8f1a4567962e3202c00faa77 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11b93c4a55ce4b36a9ef30bbfac494bb : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caa3039b9f7b490680b075df423dddfe : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b4749230c5a4fe9aa4dbb1c7e30c997 : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32059c2a270f478888702b76402f05a0 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5c63bf04e79a4fe48044959c1e7ddeae : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c43bc05ef846abb4542fd6775fd3d6 : SUCCESS in 4m 52s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/522bfc64ffb643fa9faeb2a3e5693e51 : SUCCESS in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/640e8a33b8e748e49430a6d0ffde5a96 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7982af064b4f836b05f28448dc75b78b747388b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-10-02 14:49:21.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"565934dfdb5c0ead93c76fb937774d4f16fcb022","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-10-02 14:53:46.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c793ad4c67a97a73aeed96b8cc1b31bbbcd4cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:07:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8afd7d693a9f441b83effe4fda0ef48d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/433c97778ab34f539b78051378f4c62c : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de9aa865ea234e6086896974d4b187d7 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32681c9b8765460ea21c1463c4eac407 : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/408f811a2ee142f4825be82986d02ed9 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd002c9a35304b13b0e61c52ee7f6ca3 : SUCCESS in 5m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57a18dd91f26418e917a9ca7f9a2ba94 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8b284cac4c40d4b55c232d9ede2519 : SUCCESS in 10m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/a88eb6532e5649588f9db134c58d2479 : SUCCESS in 3m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e856d3dcb50e426b86841ebdedf959fd : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1823712534c239d3f66952ff3bd28d9dbd8bb55b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-02 15:18:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0028adb079b3dbbbcea0f9f8867dcb9db8d2df34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:18:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d33da20fb49d0548f2d90793591fbb1c111588b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:56:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/921e9b3a36ee4ed782fc3803b17ab287\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1b931ec89c4a439185984d571d7635 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/058b3325c11648b3a75cb7cee63a2e9d : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a81033cbd60f4f53a6278679aae13835 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6a9982a23114072b541b3f4d6647fa1 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c165875c682a4bde9ed40e5e1da76f72 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf09ac33e3bb45b986dab786c01b1e49 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aeb3e0deea4464580f1fbc9997a08e0 : SUCCESS in 6m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/4d5fe5dc51b1412da138410b30ab9825 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"e1bb5e8ec3d6f841d14ce4f7797caadf43766f4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:56:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"026fc0186efb992f18895e989806eed411a3cc04","revisions":{"3010d1f5464d18c104e7b6f7820ae4bd38df1ee6":{"kind":"REWORK","_number":1,"created":"2024-10-01 21:32:24.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/66/931066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/931066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/931066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/931066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/931066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/931066/1"}}},"commit":{"parents":[{"commit":"8c911376e3972a9b3a834f7060ddb375c7955365","subject":"Add release note links for 2024.2 Dalmatian #4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c911376e3972a9b3a834f7060ddb375c7955365"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-10-01 21:30:42.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-10-01 21:30:42.000000000","tz":-420},"subject":"Render a graph showing the SLURP and not SLURP upgrade paths","message":"Render a graph showing the SLURP and not SLURP upgrade paths\n\nI think visual representation can help people understand the practical\neffects of the SLURP releases on their upgrade options. Lets add one\nusing graphviz to do the rendering for us.\n\nChange-Id: Ic28c8ad4bc84f094a37b38037a6859d725e85f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3010d1f5464d18c104e7b6f7820ae4bd38df1ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3010d1f5464d18c104e7b6f7820ae4bd38df1ee6"}]},"branch":"refs/heads/master"},"026fc0186efb992f18895e989806eed411a3cc04":{"kind":"REWORK","_number":2,"created":"2024-10-02 14:49:21.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/66/931066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/66/931066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/931066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/931066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/66/931066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/66/931066/2"}}},"commit":{"parents":[{"commit":"8c911376e3972a9b3a834f7060ddb375c7955365","subject":"Add release note links for 2024.2 Dalmatian #4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c911376e3972a9b3a834f7060ddb375c7955365"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-10-01 21:30:42.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-10-02 14:49:21.000000000","tz":0},"subject":"Render a graph showing the SLURP and not SLURP upgrade paths","message":"Render a graph showing the SLURP and not SLURP upgrade paths\n\nI think visual representation can help people understand the practical\neffects of the SLURP releases on their upgrade options. Lets add one\nusing graphviz to do the rendering for us.\n\nChange-Id: Ic28c8ad4bc84f094a37b38037a6859d725e85f51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/026fc0186efb992f18895e989806eed411a3cc04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/026fc0186efb992f18895e989806eed411a3cc04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
