)]}'
{"id":"openstack%2Freleases~931114","triplet_id":"openstack%2Freleases~master~I1ffcbe2d3118191ca521cd03dd967a26d371b8f6","project":"openstack/releases","branch":"master","topic":"puppet-dalmatian-release","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-10-03 15:49:16.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-10-04 10:15:54.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-04 10:28:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1ffcbe2d3118191ca521cd03dd967a26d371b8f6","subject":"Puppet OpenStack 2024.2 release","status":"MERGED","created":"2024-10-02 02:43:11.000000000","updated":"2024-10-04 10:28:05.000000000","submitted":"2024-10-04 10:28:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931114-puppet-dalmatian-release","meta_rev_id":"16a5c31aed1f1e2fde51cc4301463f545c0cbd27","_number":931114,"virtual_id_number":931114,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-04 10:28:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-10-03 15:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-10-04 10:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-10-04 10:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-10-03 13:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-02 02:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-03 10:50:24.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-10-03 12:29:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"00c18c51c60f7a81ea9ebdddfa3246b29af549b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 02:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfcad64d8adc3ea893037fe07428e3f433a460fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 02:43:40.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"75b285aaa9b10dcdaf96087817616c438f0b35c0","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 02:44:16.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"fefaa838a6ff2197af50d470f577200c0551b6e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 02:45:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"45d8d05ddbd42af88cef1a8f9e9903e8a9e30c38","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 02:45:58.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"4f7a5cdf4bb1cc4601795dd9fad145ccfb72f04b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 02:47:04.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"dafaddd474070e758586ef3f60d4ea1d981103ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 02:54:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8513c89f02fb457e8a7ac899ac6eae8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dff337bce534987a99c6da53cb03087 : SUCCESS in 6m 52s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/45cd7f57e8e5416f862a6f34035e9729 : FAILURE in 5m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0e7401a57ebf4d89a1edd604d1e9ae9c : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"edfdaded935abac2f7a15c6c10e5cdac85679e2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 03:45:00.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4c28bb34d58bd7a816d7b05111e4aa8cbc3bcc86","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 03:45:53.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"0f4b22c16b2e494d9af4c76ae91b86945f885133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 04:01:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d3f2b442f1742f59f4e41aaf5afbbf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ca30a2a9374dd89d7c5071ef0be7d1 : SUCCESS in 6m 33s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/df926a6ee0324c23adbdc15964eb753f : FAILURE in 11m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9e01508ecb4c4f3c9b106d4c444f6477 : FAILURE in 4m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9bb215df49153076c6d2e848a9b3c7ee23c6e931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 04:38:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bc18798dbc77dde2fa6f8401b9630db4cd44637f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 04:39:41.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"b8e3f1dd52359710fc60eefbddcc40bc8047e3ac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 04:40:30.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"aa734607e8e4d4aa6f071173713a0f4a92cdc391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 04:50:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c261cb26e2c491ea720af5e6a5e447f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6fa5e895baa420f80310a1b2ee98b95 : SUCCESS in 4m 44s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/09cfe9f8d794454e800d4a5706c3e48e : FAILURE in 6m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d6cb060f16ed461ca39d81aefa89bbf0 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":4},{"id":"36e1f1523cd2966256686b495b3aa22f8750c18f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 05:38:56.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8219034624040816cffd0a8f65302394670c273b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 05:39:55.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"09af1327e06d71a0d8101929991b2343c88b8473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 06:03:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26bef4e60d8c4417b67666c39891ac2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b00854d0874effb2d4cdec6a5ef3da : SUCCESS in 6m 07s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d51d157363594b9998d6559c1eb6f20b : SUCCESS in 11m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/04ceee0dfa0442b28fb9c407f08c3f12 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":5},{"id":"a1db48a91ee508d972ea2246d2adf317e8870f9f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-02 07:47:30.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"d8e89e38daccccb8f100167d5e18a871dd7f7917","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 10:23:09.000000000","message":"Patch Set 5: -Workflow","accounts_in_message":[],"_revision_number":5},{"id":"e9aa83eadd21848b35d0755acc508bac00be5aa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 10:23:12.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a53251de8b885a36b47f225cbaa6962436bd15f7","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 10:24:05.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"70196e76227d2ddebd09ea16cd50b52dd2730041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 10:41:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f817f628f6443988547c3eb31ff5bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9a50fd9a0ef48c6a532c23a0eba4a26 : SUCCESS in 5m 46s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b23f53d9670247f4a29fc8938e77750d : SUCCESS in 16m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a4703135858b47c8a2cc63c76bd917cb : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":6},{"id":"010a8dc21597b56ce9a8a3a1bf3a49c9f4e95a2a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-10-03 10:50:24.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3ffa415c2c0e0f549d5c16ac9c4d4301b813078a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-03 12:29:27.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"eeaaccfe2b3dda6e709b4fa8663b44c814224b1e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 12:39:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2598b9f3888b7f253d46824e0034d59191da6dac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 12:40:30.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"8dd853a61ba6562e601248eae3f6489aa5ba6845","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 12:41:03.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3cb478523e6cdfe7021a72f1bfac8a80462cab90","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 12:42:20.000000000","message":"Patch Set 7: PTL-Approved+1","accounts_in_message":[],"_revision_number":7},{"id":"5afe6950f6e701ab87978ded484477f9c727298e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 12:58:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10230e6d44ee48e8a0f7ecd4295dd380\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa1cb117d93043cebd53bb2528a63330 : SUCCESS in 6m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/155d2d1d41d742d1bb6d4884c21dc98f : FAILURE in 11m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/46ff140fef8b4e14848fa17e287709e6 : FAILURE in 6m 53s","accounts_in_message":[],"_revision_number":7},{"id":"8abdbfbb4a70fc36aff74dfcd33954207982333e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 13:20:57.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"7c28a4ea5c448fb9b707553bc87ca7eaa0539a4c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-03 13:21:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"de2ae082f28b166e1c4979f9b95ab75a8edae59d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:21:48.000000000","message":"Patch Set 8: PTL-Approved+1","accounts_in_message":[],"_revision_number":8},{"id":"6c0f2a1ae9cb451738bd8a40d630cfb4512fad6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 13:31:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e52dfbc1281f4b26a7daffdaa3e645ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf6de63f89e49229df76d9f9fbca64e : SUCCESS in 4m 57s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/812fa355199844daaacf1c8813dcd86c : SUCCESS in 7m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1783c43b131a4e6c8fa12c565d3c4b75 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":8},{"id":"0fe660146e833817ad39e6215ddd6c77ea2ab07b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-10-03 15:49:16.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"aebb1774cb075369d0314024bbaf506fd5be9069","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-04 10:15:54.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7be956c4d187182e6c0d7b451bdf30af65a03d39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:16:12.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5be34ac7829ec3f195adb461209c13de7d257765","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:28:04.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/891dc35cfba54c3a8849d2a21ec412cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99ee1c39c9449e1bba72cacd51b4460 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23aeac24331f40d6889ace5b58dc2dd0 : SUCCESS in 6m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/4f16bf92f8394760a50c042dfc3df5e7 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":8},{"id":"16a5c31aed1f1e2fde51cc4301463f545c0cbd27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 10:28:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"71f6c5720a4476ad52ada4c1c94144c144745ec4","revisions":{"a5f64bf7e87e1deac44ee7e51f5059041abe56e8":{"kind":"REWORK","_number":1,"created":"2024-10-02 02:43:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/1"}}},"commit":{"parents":[{"commit":"8c911376e3972a9b3a834f7060ddb375c7955365","subject":"Add release note links for 2024.2 Dalmatian #4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c911376e3972a9b3a834f7060ddb375c7955365"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5f64bf7e87e1deac44ee7e51f5059041abe56e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a5f64bf7e87e1deac44ee7e51f5059041abe56e8"}]},"branch":"refs/heads/master"},"a8aa1efba6a4d5bfb99712a8bec9012fe971ce90":{"kind":"REWORK","_number":2,"created":"2024-10-02 02:45:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/2"}}},"commit":{"parents":[{"commit":"8c911376e3972a9b3a834f7060ddb375c7955365","subject":"Add release note links for 2024.2 Dalmatian #4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c911376e3972a9b3a834f7060ddb375c7955365"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:45:26.000000000","tz":540},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8aa1efba6a4d5bfb99712a8bec9012fe971ce90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a8aa1efba6a4d5bfb99712a8bec9012fe971ce90"}]},"branch":"refs/heads/master"},"f327b602a8da74d2ba436dcff9184286f9025f39":{"kind":"REWORK","_number":3,"created":"2024-10-02 03:45:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/3"}}},"commit":{"parents":[{"commit":"8c911376e3972a9b3a834f7060ddb375c7955365","subject":"Add release note links for 2024.2 Dalmatian #4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c911376e3972a9b3a834f7060ddb375c7955365"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 03:44:51.000000000","tz":540},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f327b602a8da74d2ba436dcff9184286f9025f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f327b602a8da74d2ba436dcff9184286f9025f39"}]},"branch":"refs/heads/master"},"46583d0a63d487868dab17d1e426ea4640c5a2fe":{"kind":"REWORK","_number":4,"created":"2024-10-02 04:38:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/4"}}},"commit":{"parents":[{"commit":"8c911376e3972a9b3a834f7060ddb375c7955365","subject":"Add release note links for 2024.2 Dalmatian #4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c911376e3972a9b3a834f7060ddb375c7955365"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 04:38:44.000000000","tz":540},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46583d0a63d487868dab17d1e426ea4640c5a2fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/46583d0a63d487868dab17d1e426ea4640c5a2fe"}]},"branch":"refs/heads/master"},"5c273981687540fb5dbd9ba95c0c00ff0f43d737":{"kind":"REWORK","_number":5,"created":"2024-10-02 05:38:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/5"}}},"commit":{"parents":[{"commit":"8c911376e3972a9b3a834f7060ddb375c7955365","subject":"Add release note links for 2024.2 Dalmatian #4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8c911376e3972a9b3a834f7060ddb375c7955365"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 05:38:47.000000000","tz":540},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c273981687540fb5dbd9ba95c0c00ff0f43d737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c273981687540fb5dbd9ba95c0c00ff0f43d737"}]},"branch":"refs/heads/master"},"fe7d20ab0a25d50fd306cc38b5a99aea006cd1bf":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-10-03 10:23:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/6"}}},"commit":{"parents":[{"commit":"04107412871880811ac894bb581e5760672aa0de","subject":"Next stage for Dalmation is EOL; not UM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04107412871880811ac894bb581e5760672aa0de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 10:23:12.000000000","tz":0},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe7d20ab0a25d50fd306cc38b5a99aea006cd1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe7d20ab0a25d50fd306cc38b5a99aea006cd1bf"}]},"branch":"refs/heads/master","description":"Rebase"},"cd9e80d7ba0abd47fa4203acdfbce739468fb125":{"kind":"REWORK","_number":7,"created":"2024-10-03 12:40:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/7"}}},"commit":{"parents":[{"commit":"04107412871880811ac894bb581e5760672aa0de","subject":"Next stage for Dalmation is EOL; not UM","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/04107412871880811ac894bb581e5760672aa0de"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 12:40:30.000000000","tz":0},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd9e80d7ba0abd47fa4203acdfbce739468fb125"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cd9e80d7ba0abd47fa4203acdfbce739468fb125"}]},"branch":"refs/heads/master"},"71f6c5720a4476ad52ada4c1c94144c144745ec4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-10-03 13:20:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/931114/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931114/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931114/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931114/8"}}},"commit":{"parents":[{"commit":"865cdabf485201d96689561a11ef14ec12b47150","subject":"Merge \"Update I18n process with liaison\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/865cdabf485201d96689561a11ef14ec12b47150"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-02 02:42:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-03 13:20:57.000000000","tz":0},"subject":"Puppet OpenStack 2024.2 release","message":"Puppet OpenStack 2024.2 release\n\nChange-Id: I1ffcbe2d3118191ca521cd03dd967a26d371b8f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f6c5720a4476ad52ada4c1c94144c144745ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71f6c5720a4476ad52ada4c1c94144c144745ec4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
