)]}'
{"id":"openstack%2Freleases~931314","triplet_id":"openstack%2Freleases~master~I521640f60f39ba3ca6b75051e00bc14db143ee8c","project":"openstack/releases","branch":"master","topic":"epoxy-init-series","attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-01-31 11:12:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37702\u003e replied on the change","reason_account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2025-01-31 11:12:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37702\u003e replied on the change","reason_account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-01-31 11:12:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37702\u003e replied on the change","reason_account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}},"37702":{"account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"last_update":"2025-01-31 11:12:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37702\u003e replied on the change","reason_account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-01-31 11:12:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37702\u003e replied on the change","reason_account":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I521640f60f39ba3ca6b75051e00bc14db143ee8c","subject":"Add 2025.1 Epoxy deliverable files","status":"MERGED","created":"2024-10-03 15:46:18.000000000","updated":"2025-01-31 11:30:45.000000000","submitted":"2024-10-10 13:19:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931314-epoxy-init-series","meta_rev_id":"e786b523a59faf09c34899f00c3372173849f4e0","_number":931314,"virtual_id_number":931314,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-10 13:19:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-10-04 08:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-10-10 12:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-10-08 11:37:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-01-31 11:04:00.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":1,"date":"2024-10-04 07:28:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-10-10 12:59:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-10-03 15:50:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-03 15:47:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-03 15:59:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-04 07:28:37.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-10-04 08:02:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-10-10 12:59:14.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2025-01-31 11:04:00.000000000","updated_by":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"reviewer":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"state":"REVIEWER"},{"updated":"2025-01-31 11:30:45.000000000","updated_by":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"reviewer":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"state":"CC"}],"messages":[{"id":"f0251532c5dd32a20694d73e4662311de38c3945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-03 15:46:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ca54f7bd0836da401f5b8946e27fc4327f27630","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-03 15:50:42.000000000","message":"Patch Set 1: PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63e7e9bf909efafa4b270f42857891c3cd4e7498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 15:59:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b4762affbe4697b71d1c72ffd165b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7a485aa9494d3c9d12af9c75b18efa : SUCCESS in 7m 29s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b025f8a773564d3393832d70954cffe7 : SUCCESS in 9m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f05c1b310b124bc8bd064b7e8158d4ea : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1b1e42b77090f8a5b6ffb03dc279c5813785e0ac","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-10-04 07:28:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd0a237ede06ca684cb4cab598c3569b9d09fc99","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-10-04 08:02:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"55d1d4b1fa6c24d9a4e67230fb485ff073abadfe","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-08 11:37:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"51672acbff9b10cbbca38e1298014ce8bc8be25f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-10-10 12:59:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6f07a4f08bc4f2f11144b57f67f2668669adb0fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 12:59:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e338bf624606c48cf3515b86c7cd23919f5ad8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:19:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/220c6f5a7f574c8dbfdc2bc25ec5e437\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7040aff40c79459fa31a6f7f921e4e99 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d09db3c2b043a2a5d111420fa6c4c8 : SUCCESS in 7m 20s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/164ef06167c74f05916d048576726a06 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":1},{"id":"718a26282bed0ee725a6524676202a60317e32a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:19:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cac10d59893e094686e604c4d3edee3e587a049b","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-01-31 11:04:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"55b9fbaaec797a47434d3b66ef07f7a1efc1630c","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-01-31 11:12:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e786b523a59faf09c34899f00c3372173849f4e0","author":{"_account_id":37702,"name":"Salah Hazaa","display_name":"Salah hazaa","email":"dev.salahhazaa@gmail.com","inactive":true},"date":"2025-01-31 11:30:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86836e614fba45e396de31d58b8a582c72e7c817","revisions":{"86836e614fba45e396de31d58b8a582c72e7c817":{"kind":"REWORK","_number":1,"created":"2024-10-03 15:46:18.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/14/931314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/14/931314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/14/931314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/14/931314/1"}}},"commit":{"parents":[{"commit":"865cdabf485201d96689561a11ef14ec12b47150","subject":"Merge \"Update I18n process with liaison\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/865cdabf485201d96689561a11ef14ec12b47150"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-03 15:44:05.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-03 15:45:44.000000000","tz":120},"subject":"Add 2025.1 Epoxy deliverable files","message":"Add 2025.1 Epoxy deliverable files\n\nThe generated content initializes the 2025.1 Epoxy cycle\u0027s\ndeliverable files, based on 2024.2 Dalmatian deliverables.\n\nChange-Id: I521640f60f39ba3ca6b75051e00bc14db143ee8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/86836e614fba45e396de31d58b8a582c72e7c817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/86836e614fba45e396de31d58b8a582c72e7c817"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
