)]}'
{"id":"openstack%2Freleases~931977","triplet_id":"openstack%2Freleases~master~Ibd932d9b9e1c0f5d6e3269c6f1822d7b1d56c7be","project":"openstack/releases","branch":"master","topic":"nova-antelope-27.5.1","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2024-10-10 13:23:08.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-11-01 13:09:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd932d9b9e1c0f5d6e3269c6f1822d7b1d56c7be","subject":"[tool] Update listing tool for open and unreleased changes","status":"MERGED","created":"2024-10-09 17:08:54.000000000","updated":"2024-11-01 13:09:55.000000000","submitted":"2024-11-01 13:09:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"931977-nova-antelope-27.5.1","meta_rev_id":"37ba7cb251cbf30df913936f30c6501cf5c17b5a","_number":931977,"virtual_id_number":931977,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-01 13:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-10-15 12:38:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-11-01 12:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2024-11-01 12:56:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-10-15 12:38:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-09 17:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-10 13:05:15.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2024-11-01 12:56:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"36ac3b61b143fe21e6ac4d854d02618e7efd8171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-09 17:08:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"293a2dc24b667fa82cdf931a203f34feb7057342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-09 17:22:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b88a06d6dd04c6d9ff434121cf6c9f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1577a687d1864397bbd819e486379313 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb690105564548e3a959d2804e35fd59 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50dfc44a9d5467094fb02f0826d2a19 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03a49f626aea4bc4bd5c8195420fdb95 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a8f0136e6404e7697a3beef667dc004 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08e90a8ad7384e59b75187193322d718 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66adaab0b3304add8bd6007c421ade77 : SUCCESS in 5m 39s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/06a90b74435643efb3ae151d6e065b97 : SUCCESS in 4m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b6e84280948841ac820359efecbfa4b8 : SUCCESS in 2m 22s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9908f2eccc7f46f5bddc0715fedfc89b : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"041eec490e02a7e506614685f0b785212dd3af08","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-10-10 13:05:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"90f2a139340270c7430969a7c49c24fe0b95ec2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-10 13:10:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"21fcf3cd8e6f60f84f17fc0eebd262b41136a7fa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-10 13:11:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ec4fa211b66f5734fa8c390567c676827fc0fb4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-10-10 13:23:08.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"44c0195c03a5064b929a6abec94c271b4b54b900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 13:26:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec264c70b5d044c08cda89dcc6c71872\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35f6479e66654cc4894634251e0ccab4 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a314498dac1410b84ce44e15fa82eb1 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10444c4ebb7a43c58271e55a467403fe : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b26d54fcea2440a29a823ef83f641f96 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d4a7df180b4464588bbf2d8a94acac3 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e0203f0eb8044808f7963caf9d4fa3a : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/197b232704f947d089f96ad3a3a4a6f0 : SUCCESS in 7m 10s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c2ac7a8c8d09468aa51c1694b771f032 : SUCCESS in 6m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ed2f47803b854a469c7a8a2c5ee40f02 : SUCCESS in 3m 03s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4a12b6bad3064a6eac7501a84445534c : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9e4a66fa128ae4b9a91abaaa98baa3b816a34fae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-15 12:38:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8fc3d0b3c3c408ce543e39b9c42e6d992b20e312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 12:47:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29a79798b754397a2941340698f90b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec12ff86e3804500a3312f6fdb09ce6b : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05999c45f3904f1c9d4aa3e34b16763a : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1a2ae2dc3d84abdbfafcb285a425541 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8248150f9797418cb75bad0bf874882a : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71966448856f45e69007ddbc5e5bf21c : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d256517371e54607a9cc7c0340277ac5 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb4118aa2b64500aff75f14e877c1f3 : SUCCESS in 5m 28s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b5872c5046c341eb97ed158aede10e71 : SUCCESS in 4m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6f01982eb6f04da9bacb616e72e5c53b : SUCCESS in 3m 36s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/51e259c2b43a4db5bd37e6b50905f82e : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"7c003a79f81f4d037163a1d449eba544c7fa12fe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-01 12:56:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d8ba3aca040b0272a548293491e16d5b88bcf438","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 12:56:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a6e738a19ab20397324e8044db754605b58d730","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 13:09:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d924c85a9f5e4aec8bb4a1182d4ec89a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4445fde7cf274ba7a7d5e6e642afe865 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45bb17bf79da4cf19c7e7da2799714a5 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e762f1718dba4c0e90e7acd294c81bf0 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4779b9770cb14b11affc9c32b4f9d101 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/144b0cd0d9794b2791855e49ee5ff542 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8dab85510cd44ab8df6cfacc2e0e3fb : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b21c42d9c0040278e60e42e901955b4 : SUCCESS in 7m 50s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d5c66287601b4b30bf942b6828e36db1 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"37ba7cb251cbf30df913936f30c6501cf5c17b5a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 13:09:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6890ad92d0037696add2958bc1e34a12ea3651c3","revisions":{"ba994797e51c312c67f32d5829ffe54ed903b19f":{"kind":"REWORK","_number":1,"created":"2024-10-09 17:08:54.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/931977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/931977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/931977/1"}}},"commit":{"parents":[{"commit":"0073b3331f335cfdc96cba464b0dfd2d6c4b8fb7","subject":"Merge \"python-openstackclient 7.1.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0073b3331f335cfdc96cba464b0dfd2d6c4b8fb7"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-09 17:00:43.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-09 17:00:43.000000000","tz":120},"subject":"[tool] Update listing tool for open and unreleased changes","message":"[tool] Update listing tool for open and unreleased changes\n\nThe list_unreleased_and_open_changes.sh was not prepared to use\ndifferent branch name than the series name. Since 2023.1 Antelope\nrelease we have a \u0027series name\u0027 (antelope) and a \u0027release ID\u0027 (2023.1)\nso this patch updates the script to work properly and use the series\nname when that is required and use release ID for branch name.\n\nChange-Id: Ibd932d9b9e1c0f5d6e3269c6f1822d7b1d56c7be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba994797e51c312c67f32d5829ffe54ed903b19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba994797e51c312c67f32d5829ffe54ed903b19f"}]},"branch":"refs/heads/master"},"31fa327469a044625cb211320ff3f5fab9e66cf7":{"kind":"REWORK","_number":2,"created":"2024-10-10 13:10:17.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/931977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/931977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/931977/2"}}},"commit":{"parents":[{"commit":"0073b3331f335cfdc96cba464b0dfd2d6c4b8fb7","subject":"Merge \"python-openstackclient 7.1.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0073b3331f335cfdc96cba464b0dfd2d6c4b8fb7"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-09 17:00:43.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-10 13:10:06.000000000","tz":120},"subject":"[tool] Update listing tool for open and unreleased changes","message":"[tool] Update listing tool for open and unreleased changes\n\nThe list_unreleased_and_open_changes.sh was not prepared to use\ndifferent branch name than the series name. Since 2023.1 Antelope\nrelease we have a \u0027series name\u0027 (antelope) and a \u0027release ID\u0027 (2023.1)\nso this patch updates the script to work properly and use the series\nname when that is required and use release ID for branch name.\n\nChange-Id: Ibd932d9b9e1c0f5d6e3269c6f1822d7b1d56c7be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31fa327469a044625cb211320ff3f5fab9e66cf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/31fa327469a044625cb211320ff3f5fab9e66cf7"}]},"branch":"refs/heads/master"},"6890ad92d0037696add2958bc1e34a12ea3651c3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-15 12:38:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/77/931977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/931977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/931977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/931977/3"}}},"commit":{"parents":[{"commit":"c6ee6e9637badcb37a09a0df9bfa0f10c9420ebf","subject":"Merge \"Add releasenotes link for Tempest 2024.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6ee6e9637badcb37a09a0df9bfa0f10c9420ebf"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-09 17:00:43.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-10 13:21:56.000000000","tz":120},"subject":"[tool] Update listing tool for open and unreleased changes","message":"[tool] Update listing tool for open and unreleased changes\n\nThe list_unreleased_and_open_changes.sh was not prepared to use\ndifferent branch name than the series name. Since 2023.1 Antelope\nrelease we have a \u0027series name\u0027 (antelope) and a \u0027release ID\u0027 (2023.1)\nso this patch updates the script to work properly and use the series\nname when that is required and use release ID for branch name.\n\nChange-Id: Ibd932d9b9e1c0f5d6e3269c6f1822d7b1d56c7be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6890ad92d0037696add2958bc1e34a12ea3651c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6890ad92d0037696add2958bc1e34a12ea3651c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
