)]}'
{"id":"openstack%2Freleases~932406","triplet_id":"openstack%2Freleases~master~I117e22998b4faef3f479e2b1e4ac02e05c5e2e39","project":"openstack/releases","branch":"master","topic":"nova-antelope-27.5.1","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-10-24 07:56:20.000000000","reason":"Change was submitted"},"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-10-24 07:56:20.000000000","reason":"Change was submitted"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2024-10-22 08:42:37.000000000","reason":"\u003cGERRIT_ACCOUNT_7166\u003e replied on the change","reason_account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}},"16207":{"account":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"last_update":"2024-10-24 07:56:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I117e22998b4faef3f479e2b1e4ac02e05c5e2e39","subject":"nova: Release 2023.1 Antelope 27.5.1","status":"MERGED","created":"2024-10-15 12:38:13.000000000","updated":"2024-10-24 07:56:20.000000000","submitted":"2024-10-24 07:56:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932406-nova-antelope-27.5.1","meta_rev_id":"e104b8ac9fde4cfcf7fd5a5961f3677903e36dcd","_number":932406,"virtual_id_number":932406,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-24 07:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2024-10-24 07:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2024-10-22 09:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-10-22 08:42:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2024-10-24 07:45:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-10-22 08:43:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-15 12:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-15 12:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-15 13:20:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2024-10-15 13:20:07.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16207,"name":"ribaudr","display_name":"uggla","email":"rene.ribaud@gmail.com","username":"uggla","status":"Red Hat"},"state":"REVIEWER"},{"updated":"2024-10-15 13:20:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"state":"REVIEWER"},{"updated":"2024-10-24 07:45:08.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"e699f16a30d6b7238437df316568881a44adf62f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-15 12:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3025232b60bc34a381a0bd0071c446cc3197135e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 12:47:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d8632f283624e8884b08629afa18753\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06d0993527b4a90b6a8cb0053d41e1f : SUCCESS in 6m 22s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/8a9a6362156e456b81cd515a7bf9ee63 : SUCCESS in 7m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8d30b3a4d59a48ccb922fd837f3dc1a3 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6b251c7884a07e735d5d7e30be681c7d27904c1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-15 13:14:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2dc54b0d275ec5cf42252f7f9b166414b2531d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 13:27:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad06f67a35c24a8ea50581a1f8b9dd85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd40ba3fb464c859ff751662a631e76 : SUCCESS in 8m 44s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d3690e7965234fb3ad4dfb1efbe50ec0 : SUCCESS in 10m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d0903e52775c4dd3aafc17090e4a4256 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":2},{"id":"24e243ccbef183c1b8d8da9065c2a96e8acaa765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-22 07:16:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"06f23ac36ac2431f3225d61490d853a478cd6bb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 07:24:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f1ce7bd403474897c17fcfcc8395c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7090b1ba5fc648b6a7a00c7bdacd4c3d : SUCCESS in 5m 45s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/4d09cfc48a3447ab83f28ca54cc706fe : SUCCESS in 6m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e9e026d70ab44d4c9bb65052e82f71d7 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":3},{"id":"abb4ad4e522315517da90b66bd600e1c61fccb66","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2024-10-22 08:42:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3a16591d3e95bb89093bd810895eac48c53e9da3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 08:43:27.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"a2a1239d078a2c35140ce21d1052bb11c475371e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-10-22 09:00:08.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cf9c3438001e06c20f6be072b73a50e6f6328db","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-10-24 07:45:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"830165caa4caf92030e8cb20ed779cbbca9955e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 07:45:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"68caa400c3593998fd4fb36f1687ff4b2348bdab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 07:56:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d38de70dbad43f192f1c68638ca120d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5241cb0fa5b43afa0454b0f8cd78454 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452f6f1310904feabb48f05736980536 : SUCCESS in 5m 56s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ed9aedbf34b9470398c97048131df39f : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"e104b8ac9fde4cfcf7fd5a5961f3677903e36dcd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 07:56:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f4745285a33b32d90716503b065da8da36b05820","revisions":{"0b4ba52ad5ad7b4d039bf19c25020557d66f821f":{"kind":"REWORK","_number":1,"created":"2024-10-15 12:38:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/932406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/932406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/932406/1"}}},"commit":{"parents":[{"commit":"6890ad92d0037696add2958bc1e34a12ea3651c3","subject":"[tool] Update listing tool for open and unreleased changes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6890ad92d0037696add2958bc1e34a12ea3651c3"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-15 11:08:49.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-15 11:08:49.000000000","tz":120},"subject":"nova: Release 2023.1 Antelope 27.5.1","message":"nova: Release 2023.1 Antelope 27.5.1\n\nThis is to release latest state of stable/2023.1 before it goes to\nunmaintained/2023.1. After the transition there won\u0027t be further\nreleases for 2023.1 Antelope.\n\n$ git log --oneline --no-merges 27.5.0..HEAD\n1d788f1189 [doc]Fix the device_spec config doc about placement\ne8999f0264 Fix regression with live migration on shared storage\ncf08543ca1 hardware: Correct log\n9367b93e92 Skip snapshot test when missing qemu-img\n010acce606 libvirt: call get_capabilities() with all CPUs online\n6c3d72f477 Fix PCI passthrough cleanup on reschedule\n4ae8aa82d6 Functional test test_boot_reschedule_with_proper_pci_device_count\nfde9368dd7 Drop Fedora support\n197b14d765 [CI] Replace deprecated regex\nc534be7e00 Disconnecting volume from the compute host\nfc54bc92b5 Separate OSError with ValueError\na1fa92f0ab Added context manager for instance lock\n\nChange-Id: I117e22998b4faef3f479e2b1e4ac02e05c5e2e39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b4ba52ad5ad7b4d039bf19c25020557d66f821f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0b4ba52ad5ad7b4d039bf19c25020557d66f821f"}]},"branch":"refs/heads/master"},"0e61415ed2e2a42838ee317b7f7d0617892744dc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-15 13:14:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/932406/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/932406/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/932406/2"}}},"commit":{"parents":[{"commit":"c6ee6e9637badcb37a09a0df9bfa0f10c9420ebf","subject":"Merge \"Add releasenotes link for Tempest 2024.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6ee6e9637badcb37a09a0df9bfa0f10c9420ebf"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-15 11:08:49.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-10-15 13:14:35.000000000","tz":0},"subject":"nova: Release 2023.1 Antelope 27.5.1","message":"nova: Release 2023.1 Antelope 27.5.1\n\nThis is to release latest state of stable/2023.1 before it goes to\nunmaintained/2023.1. After the transition there won\u0027t be further\nreleases for 2023.1 Antelope.\n\n$ git log --oneline --no-merges 27.5.0..HEAD\n1d788f1189 [doc]Fix the device_spec config doc about placement\ne8999f0264 Fix regression with live migration on shared storage\ncf08543ca1 hardware: Correct log\n9367b93e92 Skip snapshot test when missing qemu-img\n010acce606 libvirt: call get_capabilities() with all CPUs online\n6c3d72f477 Fix PCI passthrough cleanup on reschedule\n4ae8aa82d6 Functional test test_boot_reschedule_with_proper_pci_device_count\nfde9368dd7 Drop Fedora support\n197b14d765 [CI] Replace deprecated regex\nc534be7e00 Disconnecting volume from the compute host\nfc54bc92b5 Separate OSError with ValueError\na1fa92f0ab Added context manager for instance lock\n\nChange-Id: I117e22998b4faef3f479e2b1e4ac02e05c5e2e39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e61415ed2e2a42838ee317b7f7d0617892744dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e61415ed2e2a42838ee317b7f7d0617892744dc"}]},"branch":"refs/heads/master","description":"Rebase"},"f4745285a33b32d90716503b065da8da36b05820":{"kind":"REWORK","_number":3,"created":"2024-10-22 07:16:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/06/932406/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/06/932406/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/06/932406/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/06/932406/3"}}},"commit":{"parents":[{"commit":"c6ee6e9637badcb37a09a0df9bfa0f10c9420ebf","subject":"Merge \"Add releasenotes link for Tempest 2024.2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c6ee6e9637badcb37a09a0df9bfa0f10c9420ebf"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-15 11:08:49.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2024-10-22 07:15:56.000000000","tz":120},"subject":"nova: Release 2023.1 Antelope 27.5.1","message":"nova: Release 2023.1 Antelope 27.5.1\n\nThis is to release latest state of stable/2023.1 before it goes to\nunmaintained/2023.1. After the transition there won\u0027t be further\nreleases for 2023.1 Antelope.\n\n$ git log --oneline --no-merges 27.5.0..HEAD\n1d788f1189 [doc]Fix the device_spec config doc about placement\ne8999f0264 Fix regression with live migration on shared storage\ncf08543ca1 hardware: Correct log\n9367b93e92 Skip snapshot test when missing qemu-img\n010acce606 libvirt: call get_capabilities() with all CPUs online\n6c3d72f477 Fix PCI passthrough cleanup on reschedule\n4ae8aa82d6 Functional test test_boot_reschedule_with_proper_pci_device_count\nfde9368dd7 Drop Fedora support\n197b14d765 [CI] Replace deprecated regex\n4c02c2d7b3 retry write_sys call on device busy\n68b9934baa add functional repoducer for bug 2065927\nc534be7e00 Disconnecting volume from the compute host\nfc54bc92b5 Separate OSError with ValueError\na1fa92f0ab Added context manager for instance lock\n75e4c86d90 database: Archive parent and child rows \"trees\" one at a time\n8823da84e9 testing: Fix and robustify archive_deleted_rows test\n\nChange-Id: I117e22998b4faef3f479e2b1e4ac02e05c5e2e39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4745285a33b32d90716503b065da8da36b05820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f4745285a33b32d90716503b065da8da36b05820"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
