)]}'
{"id":"openstack%2Freleases~933916","triplet_id":"openstack%2Freleases~master~Ia4b0b07400a6ebb9b106a09ea251c414dc37cd31","project":"openstack/releases","branch":"master","topic":"antelope-eom","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-11-07 22:25:29.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2024-11-04 19:34:19.000000000","reason":"\u003cGERRIT_ACCOUNT_9236\u003e replied on the change","reason_account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"}}},"hashtags":[],"change_id":"Ia4b0b07400a6ebb9b106a09ea251c414dc37cd31","subject":"[cinder] Transition Antelope to Unmaintained","status":"ABANDONED","created":"2024-11-01 14:46:18.000000000","updated":"2024-11-15 13:30:40.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eb7c8a7ba987bb25c3805652b733e78ed2e1b919","_number":933916,"virtual_id_number":933916,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-04 14:14:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2024-11-04 14:02:33.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 14:47:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-01 15:20:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2024-11-07 09:57:13.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"80b6031fbd46134734b953510f92724f6bc47638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-01 14:46:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30b21a4bfab5d1ce015db3936001c46de358b812","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 14:47:14.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"441fd598a553779ae5a2234c0ee7dcec3b93863a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 15:00:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c56e36fbe9984372aa188b82e5c35a66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a2ce7dcd0c4617b71b95c293137d67 : SUCCESS in 6m 04s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/331794a1fbf24c0bada39738807343bf : FAILURE in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cc90098bb64342a798f116b667906b77 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f7302a7d4e332100aa80aec8afd567fd9699992d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-01 15:13:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"711a378469edcab380b7932e562b6c56e0cb415d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-04 14:01:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ef1c44121de7358f53be3d5d6a86f144f5849ccd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 14:02:33.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a79a1350d37656ce95f1e10e1292b9303942d586","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 14:14:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67544f06e7bd4e689958169c36d81342\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12519792a6914778b1afcccb44736399 : SUCCESS in 6m 49s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/5a18d89557a34f1aa6ac07e082920e70 : FAILURE in 3m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ee00b39f180247a3aa37c0cff4856d32 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ccd77c78fde5897b4e65a9bd01ca471f7d8aed2c","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2024-11-04 19:34:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d95ca26ccc39ddacb09b7f05515d491df853cded","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-07 09:57:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd4a0de1d7fe47004b2eb38708cd1d0e74a2ad3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 10:15:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62115a24497248f0bb4dab426d1311df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b63ac5c275034628a7e29ef3613a51da : SUCCESS in 12m 52s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/65d06d356e4146a8bf557771360d6197 : FAILURE in 3m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9f5af6e0c6a44c78bf2b2d436a7cd029 : SUCCESS in 8m 18s","accounts_in_message":[],"_revision_number":2},{"id":"789547d6a2ff359e58705194d8dd29c2a9d30255","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-07 22:25:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eb7c8a7ba987bb25c3805652b733e78ed2e1b919","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-11-15 13:30:40.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/c/openstack/releases/+/934489","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"68243dcdaca8b00cf7794eccb672ef81cb460e3c","revisions":{"47e89c2c55d3ff51a6bfded2820cecd5333eefa1":{"kind":"REWORK","_number":1,"created":"2024-11-01 14:46:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/16/933916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/16/933916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/933916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/933916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/933916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/16/933916/1"}}},"commit":{"parents":[{"commit":"92e04605a3e4eb4e155fdf84baf1ba8d9e32adb3","subject":"[cinder] Final releases from stable/2023.1 (Antelope)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/92e04605a3e4eb4e155fdf84baf1ba8d9e32adb3"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-11-01 14:45:21.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-11-01 14:45:21.000000000","tz":-240},"subject":"[cinder] Transition Antelope to Unmaintained","message":"[cinder] Transition Antelope to Unmaintained\n\nThe Antelope series has reached the end of its official stable\nmaintenance period as announced in [0], so tag it as -eom and\ncreate the unmaintained/2023.1 branch from that tag.\n\nFor details about the Unmaintained state, see: [1][2].\n\n[0] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/UDQAC7SR5JAQJE5WBAG54A2MTBVBTJ44/\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#unmaintained\n[2] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ia4b0b07400a6ebb9b106a09ea251c414dc37cd31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47e89c2c55d3ff51a6bfded2820cecd5333eefa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47e89c2c55d3ff51a6bfded2820cecd5333eefa1"}]},"branch":"refs/heads/master"},"68243dcdaca8b00cf7794eccb672ef81cb460e3c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-11-04 14:01:34.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/16/933916/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/16/933916/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/933916/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/933916/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/16/933916/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/16/933916/2"}}},"commit":{"parents":[{"commit":"49b4c167e439241982f89b7f13368379bf515fdd","subject":"[cinder] Final releases from stable/2023.1 (Antelope)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/49b4c167e439241982f89b7f13368379bf515fdd"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-11-01 14:45:21.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2024-11-04 14:00:50.000000000","tz":-300},"subject":"[cinder] Transition Antelope to Unmaintained","message":"[cinder] Transition Antelope to Unmaintained\n\nThe Antelope series has reached the end of its official stable\nmaintenance period as announced in [0], so tag it as -eom and\ncreate the unmaintained/2023.1 branch from that tag.\n\nFor details about the Unmaintained state, see: [1][2].\n\n[0] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/message/UDQAC7SR5JAQJE5WBAG54A2MTBVBTJ44/\n[1] https://docs.openstack.org/project-team-guide/stable-branches.html#unmaintained\n[2] https://governance.openstack.org/tc/resolutions/20230724-unmaintained-branches.html\n\nChange-Id: Ia4b0b07400a6ebb9b106a09ea251c414dc37cd31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68243dcdaca8b00cf7794eccb672ef81cb460e3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68243dcdaca8b00cf7794eccb672ef81cb460e3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
