)]}'
{"id":"openstack%2Freleases~933944","triplet_id":"openstack%2Freleases~master~I76883ea86691e336fad51926cccfd154ffcc49f1","project":"openstack/releases","branch":"master","topic":"2023-1-last","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2025-01-16 12:42:48.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-01-16 12:55:43.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-01-16 12:55:43.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-01-14 23:01:33.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-16 12:55:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I76883ea86691e336fad51926cccfd154ffcc49f1","subject":"[neutron-tempest-plugin] Tag 2023.1-last","status":"MERGED","created":"2024-11-01 19:26:08.000000000","updated":"2025-01-16 12:55:43.000000000","submitted":"2025-01-16 12:55:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933944-2023-1-last","meta_rev_id":"6947ae224b5f62b51d62706ec1936e4712749f2b","_number":933944,"virtual_id_number":933944,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 12:55:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-01-16 12:42:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-01-15 05:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2025-01-14 23:01:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-01-16 12:42:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-01-14 23:02:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 19:26:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-01 19:29:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-01 20:20:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-11-01 20:20:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-12-19 10:31:03.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2025-01-09 13:46:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"045df1e5173d571903c8dfc950f60d9452fedfde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-01 19:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c97b1ebf942dfd466698bddf8fcb9f809d4f9ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 19:29:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b30b6e3696f4fcc9643ae8923660c23\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a792e8b34e942869e4646cb6332016e : FAILURE in 2m 31s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/cce03bd038884d17bfea3c043c3fb632 : FAILURE in 2m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a8e707c1054a486eb79ae8d4135e9df1 : FAILURE in 2m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0fedf622ba06406344b8410b30b02672dc36233b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-01 21:33:41.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2fdd3b27346c2f9d7e3590d7fee0d0fb4795b40a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:49:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8165ff48d8474b51a6305e250c7b5450\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99622fa4e0048968088e9ac5e71fff6 : SUCCESS in 5m 13s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/2b489f16543f485db8a1df91e0b7699c : FAILURE in 5m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5a7e5049cce049bfaf38c483fed7d100 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2049fcf1bb46590eff9ddeabbf499b6dfe2f06e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-10 19:11:23.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f41f3e753f0d9238924ff664b6a3d96d0c27a72e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:30:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d496eebf246414eae7dda10311258c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420dfcb5eac64581b07c6aa25338123c : SUCCESS in 7m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7af233ec588d4ddc85563e63bd44308f : SUCCESS in 6m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c2434b0d6aa64cf7a5663b5af12211ee : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":3},{"id":"c36fb7688060e73b01f5f571861cefec1e61a048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-13 15:36:10.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":4},{"id":"aac3fcc3bf22d130512b5914f94c6ec7ab31632f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 15:56:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e749c11ccfa642d0940349492350ab2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd9d3753d4b4fc78439228c9b937fe8 : SUCCESS in 7m 36s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/966aba0fbbeb491687947823fdc30af5 : SUCCESS in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ccfe716e14b1424e938bb267044177d7 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":4},{"id":"fdf53e63b95df26627563a8fe880c5473e30e61c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-12-16 08:23:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b2203606eee4e2a45963f52705700b64220909a6","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-16 08:23:48.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"71255b9548c4c287044217d3b33cbf522a001fc5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-12-16 15:07:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6948b11096910342dbcbe19d54d91a1c95e3b239","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2024-12-19 10:31:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4c21cc1af0a2c631db42a14cc55808d75a0a6321","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-09 13:46:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6512cbbcc27c236aae5ff18e0286afd351478d78","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-10 22:29:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cf8fb62e2b7e0c193234d0da9de0e87edbb06170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-10 22:29:55.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4055d2b0f554e9ba282f7a8988737e0d3fca2e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 22:45:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/693066620dc8451782e222be085799c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad80d64e6d984057a8c2da5760d83eb8 : SUCCESS in 6m 32s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d1523cce043c4d5b86f7a6b3ffcca8f1 : SUCCESS in 3m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e559005631b842a9b5711d40058d34c1 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":5},{"id":"22805143d960e00be35857d78086dd82a62130f8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-12 18:32:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1c65f193eaa03716cbb76411f8822cc805c8eefc","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-13 08:30:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d0ebb4988ed6e3f8ef20d738d47ebeacb3dca7d3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-13 15:21:41.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d079e6328fd5ccee501fcb9dc748761f7e4e2494","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 15:22:32.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"fbef9eac951108306b3ea86e276e49949705df7c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-13 15:49:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"02b4492f93681832b3ea44ab722949452d8afc03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-13 15:49:40.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"b96a9af0333112a45dc2b117de830870586123f9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 15:49:52.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"b7fda94021691726314210eedd526a76aa26321f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 16:18:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c347108aecb4db3a063418909614616\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5140a98e4d58427a87b7c4e2ecf1e3af : SUCCESS in 8m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/a9eb9786e08b4cd8b0fd1268ec230a9b : SUCCESS in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/dc8b20b43a3949d8b398bdaac946d96a : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":6},{"id":"48ed631818b7747de502c8fd039676daeb44054e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-13 19:48:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a2c4ff697011cc4467e5f14aae3b4051d64f68bc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-01-14 23:01:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"62e1872d2c33b108c4200fa7695112b8c6400c19","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 23:02:17.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"ddaaf1f632e22df741c0334a1fdb1bd518276c2a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-15 05:03:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4d2de87e7d4b047102903867aabe38c5d1b4dd3e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-01-16 12:42:48.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f95b4495a216e15e18ab5d34e6cf207e36eeace3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 12:43:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3dc26fd8a6b22405b8976f23ff6a6ad19fc5b5be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 12:55:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae033be9cec443196ab205a8a23dea7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1148dd571d4756a62fcf8452ee4bd1 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a5b4a4e4b8454faf89c7a444100ab8 : SUCCESS in 6m 43s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/aa61df58649b41a3b0a4dc53f0254aac : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":6},{"id":"6947ae224b5f62b51d62706ec1936e4712749f2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 12:55:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cdac7dac6364e146757952da48531f8bf51c57db","revisions":{"d90b03cb3016dba64cec2bd747da7ba140bf7a75":{"kind":"REWORK","_number":1,"created":"2024-11-01 19:26:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/933944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/933944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/933944/1"}}},"commit":{"parents":[{"commit":"c589ae65e9d4047856c09bf49787ccfa41e76087","subject":"Merge \"[tool] Update listing tool for open and unreleased changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c589ae65e9d4047856c09bf49787ccfa41e76087"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 19:23:36.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 19:23:36.000000000","tz":-420},"subject":"[neutron-tempest-plugin] Tag 2023.1-last","message":"[neutron-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.10.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I76883ea86691e336fad51926cccfd154ffcc49f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d90b03cb3016dba64cec2bd747da7ba140bf7a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d90b03cb3016dba64cec2bd747da7ba140bf7a75"}]},"branch":"refs/heads/master"},"60632040a1d8cb06c3fd725c549ecab0fac8a91b":{"kind":"REWORK","_number":2,"created":"2024-11-01 21:33:41.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/933944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/933944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/933944/2"}}},"commit":{"parents":[{"commit":"c589ae65e9d4047856c09bf49787ccfa41e76087","subject":"Merge \"[tool] Update listing tool for open and unreleased changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c589ae65e9d4047856c09bf49787ccfa41e76087"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 19:23:36.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-11-01 21:33:41.000000000","tz":0},"subject":"[neutron-tempest-plugin] Tag 2023.1-last","message":"[neutron-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.10.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I76883ea86691e336fad51926cccfd154ffcc49f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60632040a1d8cb06c3fd725c549ecab0fac8a91b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60632040a1d8cb06c3fd725c549ecab0fac8a91b"}]},"branch":"refs/heads/master"},"4408922d14a2bfce05d1d373ebef57912e35432c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-10 19:11:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/933944/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/933944/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/933944/3"}}},"commit":{"parents":[{"commit":"e6773524f475a91fc0f9cf9b68137b69e02c1970","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6773524f475a91fc0f9cf9b68137b69e02c1970"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 19:23:36.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-10 19:11:23.000000000","tz":0},"subject":"[neutron-tempest-plugin] Tag 2023.1-last","message":"[neutron-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.10.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I76883ea86691e336fad51926cccfd154ffcc49f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4408922d14a2bfce05d1d373ebef57912e35432c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4408922d14a2bfce05d1d373ebef57912e35432c"}]},"branch":"refs/heads/master","description":"Rebase"},"68a21e677cf7c3e90d0db5beba3482277150c35a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-12-13 15:36:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/44/933944/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/933944/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/933944/4"}}},"commit":{"parents":[{"commit":"0cff9b76cde2e764939098f9e66fe0b4190a1733","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cff9b76cde2e764939098f9e66fe0b4190a1733"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 19:23:36.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-13 15:36:10.000000000","tz":0},"subject":"[neutron-tempest-plugin] Tag 2023.1-last","message":"[neutron-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.10.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I76883ea86691e336fad51926cccfd154ffcc49f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68a21e677cf7c3e90d0db5beba3482277150c35a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/68a21e677cf7c3e90d0db5beba3482277150c35a"}]},"branch":"refs/heads/master","description":"Rebase"},"cde78faa4b9a3c05754279bb113a0b80ca08e018":{"kind":"REWORK","_number":5,"created":"2025-01-10 22:29:55.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/933944/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/933944/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/933944/5"}}},"commit":{"parents":[{"commit":"0cff9b76cde2e764939098f9e66fe0b4190a1733","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cff9b76cde2e764939098f9e66fe0b4190a1733"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 19:23:36.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2025-01-10 22:29:55.000000000","tz":0},"subject":"[neutron-tempest-plugin] Tag 2023.1-last","message":"[neutron-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.10.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I76883ea86691e336fad51926cccfd154ffcc49f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cde78faa4b9a3c05754279bb113a0b80ca08e018"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cde78faa4b9a3c05754279bb113a0b80ca08e018"}]},"branch":"refs/heads/master"},"cdac7dac6364e146757952da48531f8bf51c57db":{"kind":"REWORK","_number":6,"created":"2025-01-13 15:49:40.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/44/933944/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/44/933944/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/44/933944/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/44/933944/6"}}},"commit":{"parents":[{"commit":"0cff9b76cde2e764939098f9e66fe0b4190a1733","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cff9b76cde2e764939098f9e66fe0b4190a1733"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 19:23:36.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-01-13 15:49:24.000000000","tz":0},"subject":"[neutron-tempest-plugin] Tag 2023.1-last","message":"[neutron-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.10.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I76883ea86691e336fad51926cccfd154ffcc49f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cdac7dac6364e146757952da48531f8bf51c57db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cdac7dac6364e146757952da48531f8bf51c57db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
