)]}'
{"id":"openstack%2Freleases~933947","triplet_id":"openstack%2Freleases~master~I101a90307ef8e1787361de7e2c9ab60cb36e2151","project":"openstack/releases","branch":"master","topic":"2023-1-last","attention_set":{},"removed_from_attention_set":{"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2025-01-31 16:12:26.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-31 16:12:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I101a90307ef8e1787361de7e2c9ab60cb36e2151","subject":"[cloudkitty-tempest-tempest] Tag 2023.1-last","status":"MERGED","created":"2024-11-01 20:25:09.000000000","updated":"2025-01-31 16:12:26.000000000","submitted":"2025-01-31 16:12:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933947-2023-1-last","meta_rev_id":"bafcfaee1db092ab1dbf612a4a2472bc30f22eb4","_number":933947,"virtual_id_number":933947,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-31 16:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-01-23 13:17:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-01-20 16:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-01-23 13:17:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-01-20 16:40:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 20:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-01 20:39:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-01 21:21:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"state":"REVIEWER"},{"updated":"2025-01-20 16:40:42.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-01-23 13:17:33.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"76ef6b6a01d45f9a2dca8629e0fedefeab0b8c8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-01 20:25:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"676e64e63e6ccfc225626a0de1ded46389d0b530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 20:39:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8e8ed58d2984cca8e00fbeee4c20970\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa077158035a4fa9ac1c6dec67f43d75 : SUCCESS in 5m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/61c71222b1c14ed684355a14b5a7a730 : FAILURE in 3m 10s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8192f028933c43bebe56806ed7ad77ed : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"1f6a7a896479de5bf0c845f24b9b4603c607b6ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-10 19:11:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9014a6597ef110f96e44c63c3cd1ec3cff6c6762","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:45:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83ddb25ee99d45ecb441a2d070ff1254\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefda3235d614aea80f7730c5528cd6b : SUCCESS in 7m 19s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/3dccbe49059c4065a25463b6fd2c8e05 : SUCCESS in 6m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1121e54c614b475a9e0464b3127fc5c4 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"84907406d506225ba7f736b1498c0d8f933a51d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-13 15:36:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":3},{"id":"41ed434890aa31d9e897951cfde91a3bccc677a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 15:45:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a6468c8cf6e467687c44f5e4c0750d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa14992f4ab4e59972a25dcf3492be7 : SUCCESS in 6m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/af2c7d0919e143a2a484a88e8d457dc0 : SUCCESS in 5m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f911d45f7c9a4eeeb8487e5f4f27b7e8 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":3},{"id":"94176d1f08eae6fd71bb6523a0a08f4e8d1b6c95","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-20 16:40:42.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a643e4e2c0773e127da9d8ff8b9cc6e3a2c818d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-01-23 13:17:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8306b0309cae9ebb79825ca078f19a942fe4a07d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:18:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"baaf01b5c554d2f333ed9f5233bd3d67789a25a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 14:09:06.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/25d7b9f117964e6ca5b2a72aace2a2e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38088c279ddb400f8a4b158e41c1008d : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b71ad6e676f4c4dac0e06f3889ee6f7 : SUCCESS in 7m 48s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/18fdcca89528468e8ed456599d0ff705 : FAILURE in 6m 44s","accounts_in_message":[],"_revision_number":3},{"id":"a4f2d70fc55eaf193cc1d95ae2e9f972b3294f8c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-31 15:42:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1d6ed82f6e54991fb3b0b2812e62c3ac0883671b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:53:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bff100181a44bfabbfe8976775077e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e02b2b1a3a764fc785ba838d920daca9 : SUCCESS in 8m 41s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/3a44a934d72b43e3ac340b98b1e547f6 : SUCCESS in 6m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a51ac7898ea745708d41b83dc0ab8835 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ac91bbdf714f77c2b11615a259c43b892832e250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:54:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"12f66e0fe060dda665c2555725ed138084445bcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:12:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/962790b7ec65471a9b7749224dcad9e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8425fc545513418382d7bb88f8bf98ed : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0a2221e2894b5284d560ceb4e64626 : SUCCESS in 6m 25s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/8d4d3f0ce6274c6494e78ea90a8c0428 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"bafcfaee1db092ab1dbf612a4a2472bc30f22eb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:12:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cccbeec154e47201ff7ecfcdec543633390ee325","revisions":{"0f429ae7affa71bbbf2853a08b954662421b2a68":{"kind":"REWORK","_number":1,"created":"2024-11-01 20:25:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/933947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/933947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/933947/1"}}},"commit":{"parents":[{"commit":"c589ae65e9d4047856c09bf49787ccfa41e76087","subject":"Merge \"[tool] Update listing tool for open and unreleased changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c589ae65e9d4047856c09bf49787ccfa41e76087"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:23:55.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:23:55.000000000","tz":-420},"subject":"[cloudkitty-tempest-tempest] Tag 2023.1-last","message":"[cloudkitty-tempest-tempest] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 3.1.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I101a90307ef8e1787361de7e2c9ab60cb36e2151\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f429ae7affa71bbbf2853a08b954662421b2a68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0f429ae7affa71bbbf2853a08b954662421b2a68"}]},"branch":"refs/heads/master"},"702bd1c763a1b8944b614ba87347cc9fb757df7e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-10 19:11:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/933947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/933947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/933947/2"}}},"commit":{"parents":[{"commit":"e6773524f475a91fc0f9cf9b68137b69e02c1970","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6773524f475a91fc0f9cf9b68137b69e02c1970"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:23:55.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-10 19:11:37.000000000","tz":0},"subject":"[cloudkitty-tempest-tempest] Tag 2023.1-last","message":"[cloudkitty-tempest-tempest] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 3.1.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I101a90307ef8e1787361de7e2c9ab60cb36e2151\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/702bd1c763a1b8944b614ba87347cc9fb757df7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/702bd1c763a1b8944b614ba87347cc9fb757df7e"}]},"branch":"refs/heads/master","description":"Rebase"},"cccbeec154e47201ff7ecfcdec543633390ee325":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-13 15:36:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/47/933947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/933947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/933947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/933947/3"}}},"commit":{"parents":[{"commit":"0cff9b76cde2e764939098f9e66fe0b4190a1733","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cff9b76cde2e764939098f9e66fe0b4190a1733"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:23:55.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-13 15:36:21.000000000","tz":0},"subject":"[cloudkitty-tempest-tempest] Tag 2023.1-last","message":"[cloudkitty-tempest-tempest] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 3.1.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I101a90307ef8e1787361de7e2c9ab60cb36e2151\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cccbeec154e47201ff7ecfcdec543633390ee325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cccbeec154e47201ff7ecfcdec543633390ee325"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
