)]}'
{"id":"openstack%2Freleases~933954","triplet_id":"openstack%2Freleases~master~I74d67b7534b8ddfa474081db97c00ffcbe866d36","project":"openstack/releases","branch":"master","topic":"2023-1-last","attention_set":{},"removed_from_attention_set":{"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2025-01-31 15:59:01.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-01-20 09:39:54.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-31 15:59:01.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-01-31 15:59:01.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-01-31 15:59:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74d67b7534b8ddfa474081db97c00ffcbe866d36","subject":"[octavia-tempest-plugin] Tag 2023.1-last","status":"MERGED","created":"2024-11-01 20:51:59.000000000","updated":"2025-01-31 15:59:01.000000000","submitted":"2025-01-31 15:59:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933954-2023-1-last","meta_rev_id":"c361c0e00fd2889b936001fec6201b106c924dfd","_number":933954,"virtual_id_number":933954,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-31 15:59:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-01-23 13:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-01-20 09:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-01-23 13:16:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-01-20 09:39:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 20:52:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-01 20:57:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-01 21:20:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-11-01 21:20:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-11-01 21:20:49.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2025-01-10 19:00:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-01-23 13:16:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"5756254f4185615e9912796b9b23deeeb37b760b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-01 20:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a36bb2918f6374cbfec912a097708219c15a87a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 20:57:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abdaceb97e124738ad1be7553740095f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bbeda86a38c416c8fffb21ce92ebd38 : SUCCESS in 5m 00s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/645c590485d1459cad73d3028186968a : FAILURE in 4m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c66b566c13f7421f9ed4e9c155df0bbb : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"fe03201aa3706ee9d510ec9eb67a74872c935a91","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-14 09:00:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afffce87056a6e9c431bb14b93cd6012b24b0851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 09:08:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3cc555fd7794629b0db02c5e6e31989\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d37d71fb6a64aa68e29646deaec0fbf : SUCCESS in 7m 11s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/dbd43e16101c40c1983b669d7da3bb1d : FAILURE in 4m 38s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/703f199cdb5947cfa798a9207b2d32b0 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e948e1545208240ac5cf76e4941364e5f0fab943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-10 19:10:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"22e8ffbcfabaf06c0a4db4161476fa2323a66898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:20:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7b16954ded2412e9ae8d6af3b5831fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0178fd8deb4aae993bcee6d358eec5 : SUCCESS in 4m 53s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/93366a266af24eb2af4b9de4916f752e : SUCCESS in 3m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e0302e9f00ef404eb00912df05d0ab1f : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"c97166c847ead63851d87474d3bc301c1d148b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-13 15:35:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":3},{"id":"656ab62574d35a8f6f0993c428c11eeb9abdb606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 15:45:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60faf6446d20447bb0fe38cec08dd5ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1def752673134ecb96d8411f3e828037 : SUCCESS in 6m 38s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ce5a3f286132418199ae9575e1113e15 : SUCCESS in 3m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/822b9fc9a0cc472384fc5c58afe69308 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"610fb71a7e46a7366f950b7437633ba0b8ef2f6e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-10 10:53:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ea00beb3e0cbaada9a0bd2ecfa747730b45ece41","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 10:54:04.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"edec4f11a6c5da5bf41b949ca61ebb5ae323e15e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-10 19:00:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1e76c7618020e0e0ab504633e705c71c50db91c2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-10 22:27:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e21052123f53e736efde34db446d71aacb811bda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-01-10 22:27:51.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7d96857e9901c1c78426974bd32f5e79b91dbf96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 22:40:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a43751baa24449d79f965d9105668670\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e47f1e794c24c4c8d2ee9f860ceb205 : SUCCESS in 6m 28s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/678da219f6c940ca9c4aefad2975b93b : SUCCESS in 6m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7742e8d1323747bdae587547f5cf86ae : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":4},{"id":"76e9c2173c78905200a1a2b641f0fa0b7c4b9f42","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-20 09:39:54.000000000","message":"Patch Set 4: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e45d4c679111cf017a98fd14a4ecb7ad3f2230c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-01-23 13:16:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a668bb3882e8c4dc8759ed57fa458931a46590bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:16:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b73ad63760e8db6535099fa6a105724a995d3bc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:29:49.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8325309753d4ff99daf67d166a27a01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19988c55c0648729f6787b1a4ee53c9 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4615497d235a43cb95a120bad1f6ebaa : SUCCESS in 5m 52s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fa20eaf83b1f476b93def226e3cd460f : FAILURE in 5m 29s","accounts_in_message":[],"_revision_number":4},{"id":"abf8ad0e8dd4adec0b747bc6c81491773fa58f07","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-31 15:41:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2902f8e8414beba1b777aa6b91589cbfdf5b4050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:50:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3d2c594b8744008936ee89dadda973\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c311c89668f40d9a7ebf557a434232c : SUCCESS in 7m 00s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/28d2010055a44a5aac93cdc3fd3be985 : SUCCESS in 4m 56s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/aad55f72e93c44b7b44b34010a80128d : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":4},{"id":"f933fa0dcb422deb6e9f3ebfac1943bdd8c70a16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:50:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d4d020363fb5a77fcef99aff1b834b69774fded9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:59:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4f050a75ad4cdba898c731db2d6b9e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3b3db789ee447c88b006aa010c5b35 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16d942da69146eaa55fc2cf65fc9899 : SUCCESS in 5m 25s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/03ce0ff584dd41dbb39d913d19a82889 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":4},{"id":"c361c0e00fd2889b936001fec6201b106c924dfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:59:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ab6d81df0dd4e011e597987f7d7c4720b18f3ada","revisions":{"00574035159d4f92c49f2dce09859ebe864ba2de":{"kind":"REWORK","_number":1,"created":"2024-11-01 20:51:59.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/933954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/933954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/933954/1"}}},"commit":{"parents":[{"commit":"c589ae65e9d4047856c09bf49787ccfa41e76087","subject":"Merge \"[tool] Update listing tool for open and unreleased changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c589ae65e9d4047856c09bf49787ccfa41e76087"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:51:32.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:51:32.000000000","tz":-420},"subject":"[octavia-tempest-plugin] Tag 2023.1-last","message":"[octavia-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.9.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I74d67b7534b8ddfa474081db97c00ffcbe866d36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00574035159d4f92c49f2dce09859ebe864ba2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/00574035159d4f92c49f2dce09859ebe864ba2de"}]},"branch":"refs/heads/master"},"b8601ececbda68226ed1b0548377d5bdb5cdc95a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-10 19:10:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/933954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/933954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/933954/2"}}},"commit":{"parents":[{"commit":"e6773524f475a91fc0f9cf9b68137b69e02c1970","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6773524f475a91fc0f9cf9b68137b69e02c1970"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:51:32.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-10 19:10:45.000000000","tz":0},"subject":"[octavia-tempest-plugin] Tag 2023.1-last","message":"[octavia-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.9.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I74d67b7534b8ddfa474081db97c00ffcbe866d36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8601ececbda68226ed1b0548377d5bdb5cdc95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b8601ececbda68226ed1b0548377d5bdb5cdc95a"}]},"branch":"refs/heads/master","description":"Rebase"},"a7cfda8e01cf92c23ab9e3044e606a3940abc56c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-13 15:35:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/54/933954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/933954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/933954/3"}}},"commit":{"parents":[{"commit":"0cff9b76cde2e764939098f9e66fe0b4190a1733","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cff9b76cde2e764939098f9e66fe0b4190a1733"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:51:32.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-13 15:35:16.000000000","tz":0},"subject":"[octavia-tempest-plugin] Tag 2023.1-last","message":"[octavia-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.9.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I74d67b7534b8ddfa474081db97c00ffcbe866d36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7cfda8e01cf92c23ab9e3044e606a3940abc56c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7cfda8e01cf92c23ab9e3044e606a3940abc56c"}]},"branch":"refs/heads/master","description":"Rebase"},"ab6d81df0dd4e011e597987f7d7c4720b18f3ada":{"kind":"REWORK","_number":4,"created":"2025-01-10 22:27:51.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/54/933954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/54/933954/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/54/933954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/54/933954/4"}}},"commit":{"parents":[{"commit":"0cff9b76cde2e764939098f9e66fe0b4190a1733","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cff9b76cde2e764939098f9e66fe0b4190a1733"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 20:51:32.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2025-01-10 22:27:51.000000000","tz":0},"subject":"[octavia-tempest-plugin] Tag 2023.1-last","message":"[octavia-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 2.9.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I74d67b7534b8ddfa474081db97c00ffcbe866d36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab6d81df0dd4e011e597987f7d7c4720b18f3ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ab6d81df0dd4e011e597987f7d7c4720b18f3ada"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
