)]}'
{"id":"openstack%2Freleases~933959","triplet_id":"openstack%2Freleases~master~I9e1bf44dc2c20762f04319bad161e2bf6e2ad1a8","project":"openstack/releases","branch":"master","topic":"2023-1-last","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-11-12 12:27:25.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-01-31 16:02:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e1bf44dc2c20762f04319bad161e2bf6e2ad1a8","subject":"[watcher-tempest-plugin] Tag 2023.1-last","status":"MERGED","created":"2024-11-01 21:06:43.000000000","updated":"2025-01-31 16:02:07.000000000","submitted":"2025-01-31 16:02:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933959-2023-1-last","meta_rev_id":"4c464372fa5e9e6e9a020b8120dea424db470710","_number":933959,"virtual_id_number":933959,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-31 16:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-01-23 13:17:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-01-20 16:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-01-23 13:17:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-01-20 16:29:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-01 21:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-11-01 21:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-01 21:20:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2025-01-20 16:29:01.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-01-23 13:17:11.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"5b593edd81a4691f609b98362974bd93ae437b91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-01 21:06:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcd05531e45bcc84ae6abde1691ce8b0989c77ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:15:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd6b70b96d4248fa8881695519cbbe62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1989801fe658457e991589fe980eb571 : SUCCESS in 6m 50s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c6d22603226a401e9588a2b6eb846433 : FAILURE in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4297c133de8d4e0095da9e30abbd0f0b : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"74248b8c28d447a03892cc7580164160a1777a4e","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-12 12:27:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"198f6f31bdbcf7d03aed9506b18a83d11cd0866d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-12-10 19:10:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ba714603a8c4378af2cb09f6e409370a3e471a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 19:22:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26890aec32be48e289220b9798b583e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aace02755274e78b8affa9771a033b4 : SUCCESS in 7m 37s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/53674e5d6dbb4db089e1c74aeacceba0 : SUCCESS in 4m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f002b070ecd64d1bad065364919dd26c : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"1879cf9ace2304e0ff73a55b5d30b4f886f3a58a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-13 15:35:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":3},{"id":"f695585a4976288d460815f63d7d6e944376303f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 15:53:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a85b100d7ed341f8bab2bd295ab9a3a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43fdf4319114425ea39d6fb8ac90a3d2 : SUCCESS in 8m 24s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/cf595aefb0ba4ad7a5383ca1373e7bc8 : SUCCESS in 5m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3523a444868648bfbd5eb2b6e21370b0 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"57338ed111966486cb8ea3675a5a770fffd37747","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-20 16:29:01.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ee628f92b10872126ee9ddc920f274e0721f309","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-01-23 13:17:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"71d94bf1039d8bb5c1fc0566a676bf1b88158cae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:17:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3b602b717717911b5f77ebc31dd4ed67f2dcfe7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-23 13:50:10.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29c15ce57bf14347a10bc98843c5bf38\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/722a91654be4436986037b94c636685f : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f002274d4a1b4fca89d93ec021620f3d : SUCCESS in 5m 29s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/a24eb1227db549c59649bde714f722a6 : FAILURE in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"3baabbf81481fa7355c1b2c5947184032285d06f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-31 15:42:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1db49c57a3ea737359d8730e24789ddb4e6e1316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:53:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12b1769a4efc4dea9e5c1b56a09c558d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cdc1da07dcc4b27a062855f2b5e2782 : SUCCESS in 7m 55s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/aba22b731bb542cc9121c4295df3c53d : SUCCESS in 5m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f55d81fd67af4d868b39a60b150cc60a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"98ae8156febd33f0667d200e95a6b8e117862408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:54:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"22b6c6d77e61f102e721fb09b4c70471962cd875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:02:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741b3aa24b52414c822ad0f61c0894f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56e7daa275694294a403cd6b7a87d0b8 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3844a0344814d77b82fd7e9d7add85e : SUCCESS in 6m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/0912b8e3b8ba4611be99bccea53905f8 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":3},{"id":"4c464372fa5e9e6e9a020b8120dea424db470710","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:02:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a155119a8cc84c7579d481a443a5a16f63bf707c","revisions":{"1e4acd4be4a2a79bbb4bffb7d6f808ed97d81166":{"kind":"REWORK","_number":1,"created":"2024-11-01 21:06:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/933959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/933959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/933959/1"}}},"commit":{"parents":[{"commit":"c589ae65e9d4047856c09bf49787ccfa41e76087","subject":"Merge \"[tool] Update listing tool for open and unreleased changes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c589ae65e9d4047856c09bf49787ccfa41e76087"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 21:06:05.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 21:06:05.000000000","tz":-420},"subject":"[watcher-tempest-plugin] Tag 2023.1-last","message":"[watcher-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 3.3.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I9e1bf44dc2c20762f04319bad161e2bf6e2ad1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e4acd4be4a2a79bbb4bffb7d6f808ed97d81166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1e4acd4be4a2a79bbb4bffb7d6f808ed97d81166"}]},"branch":"refs/heads/master"},"a4441d6c171e3e1340463bb4c5860ec9bef01aab":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-10 19:10:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/933959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/933959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/933959/2"}}},"commit":{"parents":[{"commit":"e6773524f475a91fc0f9cf9b68137b69e02c1970","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e6773524f475a91fc0f9cf9b68137b69e02c1970"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 21:06:05.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-10 19:10:53.000000000","tz":0},"subject":"[watcher-tempest-plugin] Tag 2023.1-last","message":"[watcher-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 3.3.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I9e1bf44dc2c20762f04319bad161e2bf6e2ad1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4441d6c171e3e1340463bb4c5860ec9bef01aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a4441d6c171e3e1340463bb4c5860ec9bef01aab"}]},"branch":"refs/heads/master","description":"Rebase"},"a155119a8cc84c7579d481a443a5a16f63bf707c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-12-13 15:35:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/59/933959/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/59/933959/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/59/933959/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/59/933959/3"}}},"commit":{"parents":[{"commit":"0cff9b76cde2e764939098f9e66fe0b4190a1733","subject":"Set antelope/2023.1 status to unmaintained","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0cff9b76cde2e764939098f9e66fe0b4190a1733"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-01 21:06:05.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-12-13 15:35:48.000000000","tz":0},"subject":"[watcher-tempest-plugin] Tag 2023.1-last","message":"[watcher-tempest-plugin] Tag 2023.1-last\n\n2023.1 branch is moving to unmaintained state now[1]. Tempest and\ntempest plugins are branchless which means master version of\nTempest and its plugins are not guaranteed to support unmaintained\nbranched. To continue testing the unmaintained branches, Tempest\nand its plugins compatible tag[2][3] needs to be released for\ntheir testing.\n\nTagging \u00272023.1-last\u0027 with the same hash of new tag 3.3.0\nwhich mark the end of support for stable/2023.1 in Tempest.\n\n[1] https://review.opendev.org/q/topic:%22zed-unmaintained%22\n[2] https://docs.openstack.org/tempest/latest/stable_branch_support_policy.html\n[3] https://docs.openstack.org/tempest/latest/tempest_and_plugins_compatible_version_policy.html\n\nChange-Id: I9e1bf44dc2c20762f04319bad161e2bf6e2ad1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a155119a8cc84c7579d481a443a5a16f63bf707c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a155119a8cc84c7579d481a443a5a16f63bf707c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
