)]}'
{"id":"openstack%2Freleases~935983","triplet_id":"openstack%2Freleases~master~I17bfe7c3042c0ba996fd9a751467e13bb23e5db4","project":"openstack/releases","branch":"master","topic":"eom-to-eol","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-10 17:54:23.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2024-12-10 17:54:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17bfe7c3042c0ba996fd9a751467e13bb23e5db4","subject":"Transition pre 2023.1 watcher deliverables to eol","status":"MERGED","created":"2024-11-21 21:29:44.000000000","updated":"2024-12-10 17:54:23.000000000","submitted":"2024-12-10 17:54:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935983-eom-to-eol","meta_rev_id":"1264fc7c342401d042177c9671f1ef0eeb06eb7d","_number":935983,"virtual_id_number":935983,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 17:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-12-10 10:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-12-10 17:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-10 17:40:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-12-10 10:59:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2024-12-10 17:54:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-21 21:31:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-22 07:29:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-12-10 17:40:24.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"34f599e5081613f9929418be5ad86e7a8778bfe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-21 21:29:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e09a0396d5a06a87b80c7c8bc6058593157a7b9c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 21:31:56.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b1425598cd30e5fb917ae3253175ac7263a0483a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 21:43:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d18a4495d4f945388a5826d96f1fed53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bed953f72d245d58936fed3caa530e9 : SUCCESS in 4m 47s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fdb39cadc8aa4eeaac54e88f1f8f1215 : FAILURE in 3m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/26540d0b370a40a28483f647b9ecf42f : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7a2fdbfac961d6a5ecc63e71646c9ee93b0b65a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-21 21:51:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4d35c106a52eb640e6efc0308a51ea60890327b4","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 21:52:58.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"df273ee7f67092c5b598e141ccfb10d28f5730af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 22:11:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e5a4d142833431ca481a445f87debc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/187b8ff8e8d44321b4a887943b732635 : SUCCESS in 6m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/7a24283827054bfb80e2cc784f76b3f0 : FAILURE in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c6f854f2620540f3aedf61645fd64819 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c22faf38666482ccebaf31a0e7f75c5cca47455b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-21 23:08:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"83e4989a60eaf5d838a05ff36e3567fb78216a41","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-21 23:08:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"391d154cdfa8231b1aa96fc05bf6c5578f8a8c1d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:09:56.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"c639244066540d18c946c15063b7160b40fffb45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 23:16:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80f23c555fc646709f66f46317531098\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c53b1ff26bc54ffeb1f3e2f14602661f : SUCCESS in 6m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/53fabb2376ab4d96b80cfd341ea7687d : SUCCESS in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bebba8bd75ff4d94986ad0a0b01ddcea : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b4cdc3d9a2091f1d4b60ef15729a7726b682acd2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-22 07:29:07.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ccc81cdfe00b32a219a4d4015bb3280b6c470a9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"real_author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-25 09:51:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_11604\u003e\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"_revision_number":4},{"id":"87a861521e2229f5b93117774e5ed305f563d06d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 09:52:11.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"c21b1c8ab709bd9c149a0e3a0e11faca91c86e19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 10:04:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ae9cc6114b94d218d404cf2ac742366\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2822770dd5940e8859b3d3c40c69aa3 : SUCCESS in 6m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/a50875b9660841168b70518307023e23 : SUCCESS in 3m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9919afde89ab49249b4522ed37bf17d6 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":4},{"id":"28df2a4a4f886f9da959bca0378d526c36778340","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-11-25 17:37:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da9648cc3eb6826d44a7ef8ffbbb733f6bfe9650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-10 07:33:20.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8188f69242fb7df76ac418094cba45df25cc7bce","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 07:34:24.000000000","message":"Patch Set 5: -PTL-Approved","accounts_in_message":[],"_revision_number":5},{"id":"8d7d552a48419ad63cf7880d6e68d50b2abbc5e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 07:41:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/325da77622ac4cf2959b82eb7c541d84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec995a20bb44576b3be4434bb15ef8d : SUCCESS in 7m 30s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ad23ffd768a24bc8a59282a86f5a8be6 : SUCCESS in 5m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2f32067ab40544c9852c302c89c9c54c : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":5},{"id":"3b835d0168c33049fe04cbc9babb3eb05ad40903","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-10 10:59:03.000000000","message":"Patch Set 5: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c24cdc6e0f3e3215c126dcaa8f1613bfaa50561d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-12-10 17:40:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fcff9dec9c57c2443a5070885eecc440f12f7273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:40:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d859992423c14cbaa2422571a0a827653bf49587","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:54:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dafdca4423734d12bdaeebadac7c3fdb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b340d3365e0448a6b3c9158d3971def6 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84dda1955b248108bd42c1009a89015 : SUCCESS in 7m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/42337645a8be493eb243d172a91f3a1b : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":5},{"id":"1264fc7c342401d042177c9671f1ef0eeb06eb7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 17:54:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7a5bb3f401d42066da5f1649327bea072643eccc","revisions":{"a37b9f1f82fa3ca59f7fd165121835228f3e5284":{"kind":"REWORK","_number":1,"created":"2024-11-21 21:29:44.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/83/935983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/935983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/935983/1"}}},"commit":{"parents":[{"commit":"47654c44a1a1a055cfc1438aca5066876d702694","subject":"Transition zed-eom branches to EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47654c44a1a1a055cfc1438aca5066876d702694"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 21:26:33.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 21:26:33.000000000","tz":0},"subject":"Transition pre 2023.1 watcher deliverables to eol","message":"Transition pre 2023.1 watcher deliverables to eol\n\nThis change build on the proceeding patches to move\nthe pre 2023.1 branches of watcher-dashboard\nand python-watcherclient to eol.\n\nChange-Id: I17bfe7c3042c0ba996fd9a751467e13bb23e5db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a37b9f1f82fa3ca59f7fd165121835228f3e5284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a37b9f1f82fa3ca59f7fd165121835228f3e5284"}]},"branch":"refs/heads/master"},"6e5b16ccda52aa9f709019a1289b4fe2c73a4641":{"kind":"REWORK","_number":2,"created":"2024-11-21 21:51:56.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/83/935983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/935983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/935983/2"}}},"commit":{"parents":[{"commit":"47654c44a1a1a055cfc1438aca5066876d702694","subject":"Transition zed-eom branches to EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47654c44a1a1a055cfc1438aca5066876d702694"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 21:26:33.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 21:51:38.000000000","tz":0},"subject":"Transition pre 2023.1 watcher deliverables to eol","message":"Transition pre 2023.1 watcher deliverables to eol\n\nThis change build on the proceeding patches to move\nthe pre 2023.1 branches of watcher-dashboard\nand python-watcherclient to eol.\n\nChange-Id: I17bfe7c3042c0ba996fd9a751467e13bb23e5db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e5b16ccda52aa9f709019a1289b4fe2c73a4641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6e5b16ccda52aa9f709019a1289b4fe2c73a4641"}]},"branch":"refs/heads/master"},"bf91c999f790dedd40a6bd63f7320649dbfc21bd":{"kind":"REWORK","_number":3,"created":"2024-11-21 23:08:55.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/83/935983/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/935983/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/935983/3"}}},"commit":{"parents":[{"commit":"47654c44a1a1a055cfc1438aca5066876d702694","subject":"Transition zed-eom branches to EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/47654c44a1a1a055cfc1438aca5066876d702694"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 21:26:33.000000000","tz":0},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 23:08:43.000000000","tz":0},"subject":"Transition pre 2023.1 watcher deliverables to eol","message":"Transition pre 2023.1 watcher deliverables to eol\n\nThis change build on the proceeding patches to move\nthe pre 2023.1 branches of watcher-dashboard\nand python-watcherclient to eol.\n\nChange-Id: I17bfe7c3042c0ba996fd9a751467e13bb23e5db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf91c999f790dedd40a6bd63f7320649dbfc21bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bf91c999f790dedd40a6bd63f7320649dbfc21bd"}]},"branch":"refs/heads/master"},"2de630d6c3836c1d30c148b00c809cb3478a3fad":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-11-25 09:51:02.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"real_uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/83/935983/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/935983/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/935983/4"}}},"commit":{"parents":[{"commit":"9d761f96db05090430928251aad4c7ad7497ca83","subject":"Transition zed-eom branches to EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9d761f96db05090430928251aad4c7ad7497ca83"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 21:26:33.000000000","tz":0},"committer":{"name":"sean mooney","email":"work@seanmooney.info","date":"2024-11-25 09:51:02.000000000","tz":0},"subject":"Transition pre 2023.1 watcher deliverables to eol","message":"Transition pre 2023.1 watcher deliverables to eol\n\nThis change build on the proceeding patches to move\nthe pre 2023.1 branches of watcher-dashboard\nand python-watcherclient to eol.\n\nChange-Id: I17bfe7c3042c0ba996fd9a751467e13bb23e5db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2de630d6c3836c1d30c148b00c809cb3478a3fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2de630d6c3836c1d30c148b00c809cb3478a3fad"}]},"branch":"refs/heads/master","description":"Rebase"},"7a5bb3f401d42066da5f1649327bea072643eccc":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-12-10 07:33:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/83/935983/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/83/935983/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/83/935983/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/83/935983/5"}}},"commit":{"parents":[{"commit":"7acb98f24f798d7ff708f6d47dd7973a4245bb41","subject":"Transition zed-eom branches to EOL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7acb98f24f798d7ff708f6d47dd7973a4245bb41"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2024-11-21 21:26:33.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-12-09 22:25:17.000000000","tz":60},"subject":"Transition pre 2023.1 watcher deliverables to eol","message":"Transition pre 2023.1 watcher deliverables to eol\n\nThis change build on the proceeding patches to move\nthe pre 2023.1 branches of watcher-dashboard\nand python-watcherclient to eol.\n\nChange-Id: I17bfe7c3042c0ba996fd9a751467e13bb23e5db4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a5bb3f401d42066da5f1649327bea072643eccc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7a5bb3f401d42066da5f1649327bea072643eccc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
