)]}'
{"id":"openstack%2Freleases~938231","triplet_id":"openstack%2Freleases~master~I1aa7221e984e5005a5a01b31bc5806d1c42918f3","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"3009":{"account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"last_update":"2025-01-08 20:53:15.000000000","reason":"\u003cGERRIT_ACCOUNT_3009\u003e replied on the change","reason_account":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"}}},"hashtags":[],"change_id":"I1aa7221e984e5005a5a01b31bc5806d1c42918f3","subject":"OpenStack-Helm 2024.2.0","status":"ABANDONED","created":"2024-12-24 14:28:24.000000000","updated":"2025-01-09 20:20:14.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0ba235c0f20e1e2846be0439b06b787cb3041df","_number":938231,"virtual_id_number":938231,"owner":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-01-06 23:54:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"all":[{"value":1,"date":"2025-01-07 15:10:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"all":[{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"value":-1,"date":"2025-01-08 20:53:15.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-01-06 23:46:06.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-24 14:29:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-24 15:06:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 15:10:46.000000000","updated_by":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"reviewer":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"state":"REVIEWER"},{"updated":"2025-01-08 09:41:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"5b62cc954f602267ba2228b47e20eb734fccd000","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2024-12-24 14:28:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e807446c5aee2be3fc1458d1876dcd554ec81a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-24 15:06:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e543c433d3c14591a6909780cc799620\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d52cc642d714b5b912a8eaaa36519ec : SUCCESS in 5m 22s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/add4cc47b3ff49bb9aac692c455a8f68 : FAILURE in 4m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6a5bed3a4c1e4456a89656435859aa07 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":1},{"id":"2105354ad4dbb29a536856aa64dfb5e6edae6e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-02 17:36:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c63477d3f041418d1e2d1b9d13ce517166bc0faf","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 17:37:38.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"9ff4ce2cd40ec7f58c203a886b9ee8bbf9cd88f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-02 17:48:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3337343dd1f64789ace0b24d8e52bdd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/799e5399b5d049fea17f47a6d4f2d548 : SUCCESS in 6m 08s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/748c978e1a444e0b9f462a6887ef0087 : FAILURE in 5m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f8105165e9eb40af83c81d7211d8437f : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ed2d0a5bb22c529b683e0922c5b0dae0b1af55f1","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-06 12:43:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3e09dccccb0a43d9ff36baec264a7533523a4ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 12:50:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdc54c065af346949cf1d94cbadd01ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9c42efb3e649d2b5447526a4c5a6d3 : SUCCESS in 6m 20s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/04fed2bed54a4df4a1562a68008b8d11 : FAILURE in 4m 53s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/85584faec669467e9758e531189a8e56 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5f7df380ff078889583419bc5cdf5c500ad1d696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-06 23:45:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d5bb3a3c7ce55776e7d578660460bb4de7a2cc41","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 23:46:06.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"387d18a501f5dca8eae38037becf9fd5bca07302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 23:54:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73c4ce9bf244588bcd066bc63f91e86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726970856193486586c6dc6c2f10eed5 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6af55da36ac240629eb571d93f626d9a : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fe023ed32824684b069884583d1f2df : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4264886bcb914f19a9fbc108cf244388 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b88a1ad1e1ed4f779c4998edd53db8dd : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/027285571aad4b418c4d00f6227e7166 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10662e8d9b594eea85d4fe9f2d5efd14 : SUCCESS in 5m 11s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/70127412160946e59fff5aad7d0e5663 : SUCCESS in 3m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9a1ed108d0f1490eac6026e6fe4bedd6 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":3},{"id":"2f2b9c8a353b0067716c7a70deeaf3889f67632f","author":{"_account_id":21040,"name":"Alexey","email":"aodinokov@mirantis.com","username":"aodinokov"},"date":"2025-01-07 15:10:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7c72305021c72ea0509ce0ddaed34976010bbdc5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-08 09:41:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ae78abc3d3de3132eb49a4e847ed68ae1b55d81c","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-08 20:53:15.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"fe3ceebee1166bb559be51e088b1f31061e7ac4e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-09 20:12:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"c0ba235c0f20e1e2846be0439b06b787cb3041df","author":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"date":"2025-01-09 20:20:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6f58ac50bab20878ffc9c51330a7ed07fa09e860","revisions":{"4ea900153b4640cde177eeb11ee9688b950b74e1":{"kind":"REWORK","_number":1,"created":"2024-12-24 14:28:24.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/938231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/938231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/938231/1"}}},"commit":{"parents":[{"commit":"53e5eaa269d26eebb698bb08ab9128a7758e3b2b","subject":"EOL Ansible-Collection-Kolla Yoga and Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53e5eaa269d26eebb698bb08ab9128a7758e3b2b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-12-24 14:26:35.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-12-24 14:28:12.000000000","tz":-360},"subject":"OpenStack-Helm 2024.2.0","message":"OpenStack-Helm 2024.2.0\n\nChange-Id: I1aa7221e984e5005a5a01b31bc5806d1c42918f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ea900153b4640cde177eeb11ee9688b950b74e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ea900153b4640cde177eeb11ee9688b950b74e1"}]},"branch":"refs/heads/master"},"ceacf0c79cefb95a7f32691a59ef8e29b07c79ca":{"kind":"REWORK","_number":2,"created":"2025-01-02 17:36:53.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/938231/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/938231/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/938231/2"}}},"commit":{"parents":[{"commit":"53e5eaa269d26eebb698bb08ab9128a7758e3b2b","subject":"EOL Ansible-Collection-Kolla Yoga and Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53e5eaa269d26eebb698bb08ab9128a7758e3b2b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-12-24 14:26:35.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-02 17:36:34.000000000","tz":-360},"subject":"OpenStack-Helm 2024.2.0","message":"OpenStack-Helm 2024.2.0\n\nChange-Id: I1aa7221e984e5005a5a01b31bc5806d1c42918f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ceacf0c79cefb95a7f32691a59ef8e29b07c79ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ceacf0c79cefb95a7f32691a59ef8e29b07c79ca"}]},"branch":"refs/heads/master"},"6f58ac50bab20878ffc9c51330a7ed07fa09e860":{"kind":"REWORK","_number":3,"created":"2025-01-06 23:45:18.000000000","uploader":{"_account_id":3009,"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","username":"kozhukalov"},"ref":"refs/changes/31/938231/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/938231/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/938231/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/938231/3"}}},"commit":{"parents":[{"commit":"53e5eaa269d26eebb698bb08ab9128a7758e3b2b","subject":"EOL Ansible-Collection-Kolla Yoga and Zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53e5eaa269d26eebb698bb08ab9128a7758e3b2b"}]}],"author":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2024-12-24 14:26:35.000000000","tz":-360},"committer":{"name":"Vladimir Kozhukalov","email":"kozhukalov@gmail.com","date":"2025-01-06 23:45:07.000000000","tz":-360},"subject":"OpenStack-Helm 2024.2.0","message":"OpenStack-Helm 2024.2.0\n\nChange-Id: I1aa7221e984e5005a5a01b31bc5806d1c42918f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f58ac50bab20878ffc9c51330a7ed07fa09e860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f58ac50bab20878ffc9c51330a7ed07fa09e860"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
