)]}'
{"id":"openstack%2Freleases~938580","triplet_id":"openstack%2Freleases~master~Icc1df7e3fa512f504c6c378883bdc92c92d5287e","project":"openstack/releases","branch":"master","topic":"epoxy-milestone-2","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-01-16 13:32:09.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-10 10:05:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-01-16 13:32:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icc1df7e3fa512f504c6c378883bdc92c92d5287e","subject":"Release oslo.concurrency for Epoxy-2 milestone","status":"MERGED","created":"2025-01-07 16:36:07.000000000","updated":"2025-01-16 13:32:09.000000000","submitted":"2025-01-16 13:32:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938580-epoxy-milestone-2","meta_rev_id":"98a155cb02ab0a1ae1da287bbebb9864f1c39652","_number":938580,"virtual_id_number":938580,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 13:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-01-16 13:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-01-10 13:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-01-10 15:17:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":1,"date":"2025-01-10 10:05:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-01-16 13:15:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-01-10 10:06:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-07 16:36:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-07 16:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 17:11:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-01-07 17:11:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-10 15:17:16.000000000","updated_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"reviewer":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"state":"REVIEWER"},{"updated":"2025-01-16 13:15:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"1d86fb3e7756df9e61b37756d7ff671f0d89a879","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-07 16:36:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e4e0377cab447b457c450b3a3ecbfd32b522230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 16:58:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dd262806f7f4449b1a44d07318d86ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12038e1b27114ff093890cb048d5a6c7 : SUCCESS in 6m 55s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c510418277ab427aa29af49c9f7c8771 : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8c1faca61e2045739d890298b4ef84de : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3e21ae530ce5fd92192d9cac7c68435a133030dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-01-10 10:05:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e911b960825ebf8d97835352c8e7f0086fc2c10f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 10:06:14.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"715d66d7708d1836b3eec045da1005f3434824f5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-01-10 13:47:47.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25e59a0df08bce433b2d668c8ff241308f23528f","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2025-01-10 15:17:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f4b000fc6fa882d0ac52e9380f438fa62f478d7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-01-16 13:15:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ccdf4c05a93871639c3eaf3e729a779fed74bc7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:16:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef15a68251aee8db0127250aea8eef354943c698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:32:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee8e9b40e64f463ba369fe3c6f61e572\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe28eaaddf134488bda8d2499743b315 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2190280b864cc6abffcc08e5681994 : SUCCESS in 6m 44s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/b9208d7bcd2248a98fc43477ce691e34 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"98a155cb02ab0a1ae1da287bbebb9864f1c39652","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:32:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"909c3dcbc63fd60b17fe754799eddc560df88dbd","revisions":{"909c3dcbc63fd60b17fe754799eddc560df88dbd":{"kind":"REWORK","_number":1,"created":"2025-01-07 16:36:07.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/80/938580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/80/938580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/938580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/938580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/80/938580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/80/938580/1"}}},"commit":{"parents":[{"commit":"0fe5f6ea42562bae373b9a48b3f8757505c4275b","subject":"Merge \"[mistral-tempest-plugin] Tag 2023.1-last\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0fe5f6ea42562bae373b9a48b3f8757505c4275b"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-01-07 16:36:03.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-01-07 16:36:03.000000000","tz":60},"subject":"Release oslo.concurrency for Epoxy-2 milestone","message":"Release oslo.concurrency for Epoxy-2 milestone\n\nThis is the Epoxy-2 milestone release for oslo.concurrency.\nClients and libraries following the cycle-with-intermediary release\nmodel are encouraged to release each milestone to make sure updates\nare made available.\n\nIf the team is happy with the current release hash, please +1 this\npatch and we will process the release. If you need to get some\nimportant changes merged first, leave a -1 and update the patch with\na new commit hash to use once it is ready.\n\nIf a release is definitely not wanted at this time, please -1 and\nleave a comment explaining that and we can abandon the patch.\n\nAny patches with no response will be assumed to be OK and will be\nprocessed after January 9th.\n\nChange-Id: Icc1df7e3fa512f504c6c378883bdc92c92d5287e\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/909c3dcbc63fd60b17fe754799eddc560df88dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/909c3dcbc63fd60b17fe754799eddc560df88dbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
