)]}'
{"id":"openstack%2Freleases~940118","triplet_id":"openstack%2Freleases~master~I26b10ec48ebd46e86516a88696b7e451aff0fd95","project":"openstack/releases","branch":"master","topic":"pbr","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-02-04 14:20:43.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-02-04 14:20:43.000000000","reason":"Change was submitted"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-02-03 19:08:53.000000000","reason":"\u003cGERRIT_ACCOUNT_31245\u003e replied on the change","reason_account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-02-04 14:20:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26b10ec48ebd46e86516a88696b7e451aff0fd95","subject":"pbr 6.1.1","status":"MERGED","created":"2025-01-24 10:30:56.000000000","updated":"2025-02-04 14:20:43.000000000","submitted":"2025-02-04 14:20:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940118-pbr","meta_rev_id":"d32196abc3be8313c8c8ba658cc1cab503a9839f","_number":940118,"virtual_id_number":940118,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-04 14:20:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-04 13:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-02-04 07:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-01-29 10:14:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-03 19:08:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":1,"date":"2025-02-04 14:16:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},{"value":1,"date":"2025-01-28 20:25:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-04 13:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-02-03 19:09:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"}],"CC":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 10:31:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-24 10:54:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-24 12:01:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-01-24 12:01:52.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-01-24 12:06:10.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-01-28 14:28:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2025-01-30 13:23:23.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"CC"},{"updated":"2025-02-04 07:26:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-02-04 13:47:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-02-04 14:16:03.000000000","updated_by":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"reviewer":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"state":"REVIEWER"}],"messages":[{"id":"8edbf8e28e6580fb81fc2be3aef734a5ffdea60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-24 10:30:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fae98d2fda00623d271eac43af5cc638a4b2d460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 10:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/304aa7382a074ec1ab5dca1eeb647c62\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2e4ebff353f408cbd67c53dbc7658fb : SUCCESS in 6m 49s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/19912072d0d544e481f80261485dcd36 : SUCCESS in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/08fe61d17a12482bb0cb0e2a4d1a245a : SUCCESS in 8m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5843b9ed04c8b9d29d47caa0b78f85a202c7ad22","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-24 12:06:10.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a398476f02c0f6ec9241d1e0590ccecf4aba500","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 14:28:51.000000000","message":"Patch Set 1: Code-Review+1\n\nThis has been exercised in a number of configurations via I2f7940e14a1d18dec13a15e3ef66af0381e515f5 as well as I0c2be26e97cbe6c8f9e78de88e27b5e2a17af832 so should be fine to proceed if others are ready.\n\nI\u0027m unclear why this is 6.2.0 instead of 6.1.1, but am okay with either.","accounts_in_message":[],"_revision_number":1},{"id":"50efefd1a9b69a947043f772a40c4d7cd6a06467","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 14:30:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de07cc96f3663dd1c68f398332f79c8b601706c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 16:10:08.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b6edcfcf549ffbf30954fae53b01c61dff7bbfea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 16:10:23.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/releases for 940118,2","accounts_in_message":[],"_revision_number":2},{"id":"eb26c0eaeea90fc03f4759d8253ea7dbcae34429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 16:11:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bf399f54d9bb0ac7d4403d98330f636329cba265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 16:25:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1b9ce2510be4db1bce955f66d2f87de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf3ca6b533243288d2808bc5d6bc9f5 : SUCCESS in 6m 18s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/3d96c124d10c492e87ea6e8179090b56 : SUCCESS in 6m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2630dd5b712c4fa384f077afc93daed5 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":3},{"id":"81d5933426cb5dfd91c547d304462f5a51b21616","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 17:25:46.000000000","message":"Patch Set 3:\n\nI\u0027d really like to get Ia3b6d55b0e53361a73a16b5c861d8fb04dfd0507 included as soon as it merges, since it\u0027s become apparent that will enable us to simplify the use of pbr.build in pyproject.toml files (without it, projects will still have to explicitly list SetupTools when building with pyproject-build on pre-3.12 Pythons, because build omits and uninstalls SetupTools from the base set regardless of Python version).","accounts_in_message":[],"_revision_number":3},{"id":"4ba54ee627b1d3c5f072063df7c39885f311c789","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 17:34:03.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ad3f81c7c870d708982e4d70abcf057529bff5d2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 17:35:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8973ff96afbb46f17fc44d8fa6248b85967faeea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 20:24:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"521a11ed446510642c1d190ff961fda80ddd38a4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-01-28 20:25:26.000000000","message":"Patch Set 4: Code-Review+1\n\nRevised to include the indicated changes now that they\u0027ve also merged.","accounts_in_message":[],"_revision_number":4},{"id":"423987e167a7c1566fb218977a3b9fc0f7b818ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 20:31:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90f7aeaaec9e41519bfa9ccf706eb892\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5114acc1a4e408794bd567f7a89611d : SUCCESS in 5m 51s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/2b27527171674d6da6f6dc977cf8fdd9 : SUCCESS in 6m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/25b2b98d10df448ab548d63bf25d1327 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":4},{"id":"607a17c616c5729b01e55486f411e84483fd6706","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-29 10:14:05.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9e4773db3c15bb37e06f3875d59668ffe0f4d735","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-01-30 13:23:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aaf9e7bc42e9292d5b4424e67d4a8f580f4e4055","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2025-02-03 19:08:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"406fd350be07ed9b8bee681de3b3e0fd8a1b523c","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 19:09:36.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"0f20d357c71985373b326c87b103a1a5c7a0d1da","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-04 07:26:52.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0038d12ef8f130a4acf3fc675f2a37479ea25d82","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-04 13:47:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2924e2d2dfcfe50503cb8b6fbcac918f84a94709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 13:47:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"622c9d93a78c3361eafa64570ae56072d6955c59","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"date":"2025-02-04 14:16:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"776ea2aa1740e4c533fb4f965963bf6538f0a366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 14:20:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e708064af46413c819a00036f66f32b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8aedb51ffb42ed979e4841da44cd75 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dea85a5c7dd242f98918a4496ea59eec : SUCCESS in 7m 02s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/32ae98784fb64676a3c3aa5b7894786f : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"d32196abc3be8313c8c8ba658cc1cab503a9839f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 14:20:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"580e87649e00b909f957d66e5077bffe5c708049","revisions":{"58d971cab9c40c668bdc1cec87906da209164d5d":{"kind":"REWORK","_number":1,"created":"2025-01-24 10:30:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/940118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/940118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/940118/1"}}},"commit":{"parents":[{"commit":"fc8986c4e694b2a5751337602a3209d0362293bd","subject":"Merge \"[Puppet OpenStack] Transition 2023.1 Antelope to Unmaintained\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc8986c4e694b2a5751337602a3209d0362293bd"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:30:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:30:46.000000000","tz":0},"subject":"pbr 6.2.0","message":"pbr 6.2.0\n\nMake setuptools a dependency since it is no longer installed by default.\n\nChange-Id: I26b10ec48ebd46e86516a88696b7e451aff0fd95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58d971cab9c40c668bdc1cec87906da209164d5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/58d971cab9c40c668bdc1cec87906da209164d5d"}]},"branch":"refs/heads/master"},"b36eb509968033c63efb13be76199ed068359103":{"kind":"REWORK","_number":2,"created":"2025-01-28 16:10:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/940118/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/940118/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/940118/2"}}},"commit":{"parents":[{"commit":"fc8986c4e694b2a5751337602a3209d0362293bd","subject":"Merge \"[Puppet OpenStack] Transition 2023.1 Antelope to Unmaintained\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fc8986c4e694b2a5751337602a3209d0362293bd"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:30:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-28 16:10:08.000000000","tz":0},"subject":"pbr 6.1.1","message":"pbr 6.1.1\n\nMake setuptools a dependency since it is no longer installed by default.\n\nChange-Id: I26b10ec48ebd46e86516a88696b7e451aff0fd95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b36eb509968033c63efb13be76199ed068359103"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b36eb509968033c63efb13be76199ed068359103"}]},"branch":"refs/heads/master"},"50c8cba7b7644021631cbebb5a5843a9254b58b0":{"kind":"REWORK","_number":3,"created":"2025-01-28 16:11:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/18/940118/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/940118/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/940118/3"}}},"commit":{"parents":[{"commit":"501d95e0667f81018ca8bc47b27108d4d3605968","subject":"Merge \"neutron-lib: Release 3.18.0 for epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/501d95e0667f81018ca8bc47b27108d4d3605968"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:30:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-28 16:11:41.000000000","tz":0},"subject":"pbr 6.1.1","message":"pbr 6.1.1\n\nMake setuptools a dependency since it is no longer installed by default.\n\nChange-Id: I26b10ec48ebd46e86516a88696b7e451aff0fd95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/50c8cba7b7644021631cbebb5a5843a9254b58b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/50c8cba7b7644021631cbebb5a5843a9254b58b0"}]},"branch":"refs/heads/master"},"580e87649e00b909f957d66e5077bffe5c708049":{"kind":"REWORK","_number":4,"created":"2025-01-28 20:24:21.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/18/940118/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/940118/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/940118/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/940118/4"}}},"commit":{"parents":[{"commit":"501d95e0667f81018ca8bc47b27108d4d3605968","subject":"Merge \"neutron-lib: Release 3.18.0 for epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/501d95e0667f81018ca8bc47b27108d4d3605968"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-01-24 10:30:42.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-01-28 20:24:01.000000000","tz":0},"subject":"pbr 6.1.1","message":"pbr 6.1.1\n\nMake setuptools a dependency since it is no longer installed by default.\n\nChange-Id: I26b10ec48ebd46e86516a88696b7e451aff0fd95\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/580e87649e00b909f957d66e5077bffe5c708049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/580e87649e00b909f957d66e5077bffe5c708049"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
