)]}'
{"id":"openstack%2Freleases~941455","triplet_id":"openstack%2Freleases~master~I7cd5e888f56d4cf7f2d7ec0acb70618355d68286","project":"openstack/releases","branch":"master","topic":"bobcat-stable","attention_set":{},"removed_from_attention_set":{"6732":{"account":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"last_update":"2025-05-01 08:09:53.000000000","reason":"Change was abandoned"},"26285":{"account":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"last_update":"2025-05-01 08:09:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7cd5e888f56d4cf7f2d7ec0acb70618355d68286","subject":"Release trove for stable/2023.2","status":"ABANDONED","created":"2025-02-12 15:00:17.000000000","updated":"2025-05-01 08:09:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bfa740748f9f38ee63221eed0acb4177ecd4fe40","_number":941455,"virtual_id_number":941455,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-12 15:10:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-12 15:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-12 15:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 15:50:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2025-02-12 15:50:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"}],"messages":[{"id":"6421b056ff0acd8c7b6f844ed5b101f7144fd18f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-02-12 15:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1973f2c0da070168ef1732fc4008b7cc6d16fc07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 15:10:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/732fb3c4c9f64fe48a1f41d958e383ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea94627a3e44aa38ffccc8cc8a3ee9e : SUCCESS in 6m 34s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1737b9d17a704e0ea51a3ec8f0230251 : SUCCESS in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c7c9e8c2193c488f92c1756239938fde : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bfa740748f9f38ee63221eed0acb4177ecd4fe40","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-01 08:09:53.000000000","message":"Abandoned\n\nno response from team, so let\u0027s just abandon this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b4c7c1e7fb6a0f5f2b314f772f01c519caa4fd6","revisions":{"6b4c7c1e7fb6a0f5f2b314f772f01c519caa4fd6":{"kind":"REWORK","_number":1,"created":"2025-02-12 15:00:17.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/55/941455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/941455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/941455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/941455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/941455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/941455/1"}}},"commit":{"parents":[{"commit":"1a780b4fe439b85f1ebc0758e14f0ecf2293df66","subject":"Merge \"Move 2023.2 Bobcat\u0027s EOL date away from Epoxy release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1a780b4fe439b85f1ebc0758e14f0ecf2293df66"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-12 15:00:12.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-02-12 15:00:12.000000000","tz":60},"subject":"Release trove for stable/2023.2","message":"Release trove for stable/2023.2\n\nAs a preparation for the transition of 2023.2 Bobcat to End of Life [1],\nthis release picks up new commits to trove since\nthe last release from stable/2023.2.\n\nThis is being proposed as a convenience to help make sure stable\nchanges are being released. If the team is good with this going out,\nplease respond with a +1 to let the release team know it is OK to\nproceed.\n\nIf it is not wanted at this time, or if there are more changes that\nwould be good to get merged before doing a stable release, please\nleave a -1 with a comment with what the team would prefer. We can\nthen either abandon this patch, or wait for an update with a new\ncommit hash to use instead.\n\n[1] https://releases.openstack.org/\n\n$ git log --oneline --no-merges 20.0.0..a391f6c4\na391f6c4 [CI]: run perodical jobs on master branch only\nfc8cacef Change settings for stable/2023.2\n9e9f8b70 Enable network_isolation by default\n16e0f953 Fix error can\u0027t detach the configuration group binding\nbc0ee003 Retire TripleO: remove the TripleO jobs/ref\n2191f55b Fix potential network conflict\n7a039ecb stable-only: Pin bandit\n9437bb4a Don\u0027t configure ip route for slaac ipv6 subnet\n0fd72e89 Fix guest agent failed to use swiftclient over TLS\n505a9834 Add region name to trove-guestagent\n25eb9d54 Fixes the way to check db instance status\n446e40e4 Update TOX_CONSTRAINTS_FILE for stable/2023.2\n73d553f6 Update .gitreview for stable/2023.2\n\nChange-Id: I7cd5e888f56d4cf7f2d7ec0acb70618355d68286\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b4c7c1e7fb6a0f5f2b314f772f01c519caa4fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b4c7c1e7fb6a0f5f2b314f772f01c519caa4fd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
