)]}'
{"id":"openstack%2Freleases~942918","triplet_id":"openstack%2Freleases~master~I5505bf939cbe464d2e89908bf648c5bc8dda7b8a","project":"openstack/releases","branch":"master","attention_set":{"37710":{"account":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"last_update":"2025-03-18 11:42:07.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-18 10:46:47.000000000","reason":"Change was abandoned"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-03-18 10:46:47.000000000","reason":"Change was abandoned"},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-03-18 10:46:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5505bf939cbe464d2e89908bf648c5bc8dda7b8a","subject":"Release openstackdocstheme 3.5.0","status":"ABANDONED","created":"2025-02-27 15:58:09.000000000","updated":"2025-03-18 11:42:07.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"de50d9230a270008d756603006fd6e83dfa58843","_number":942918,"virtual_id_number":942918,"owner":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-09 09:02:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"disliked":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":-1,"date":"2025-03-10 19:01:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-03-09 08:58:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-03-09 08:59:05.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-27 15:58:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-27 16:19:19.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-02-27 16:42:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-02-27 20:45:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-01 04:22:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-03-08 11:44:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-03-10 19:01:12.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4eac0a9151991197826d45c82c471e419d1dba62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"date":"2025-02-27 15:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6aebabeb9746a59aed781b6a1a24391d689d835b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-27 16:19:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"462fd04cd3dbcb56832117266597943b1c52a499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"date":"2025-02-27 16:40:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"f3831802f60a50c38dbf2c5c25088813720d402f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 20:45:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c91be39ad124d9095fa3c5b688ea68e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936e6a3fc2b44b9cbfa2d78fca2b8688 : SUCCESS in 7m 44s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/aa3c0a20120643b0b2a2757af907cdd1 : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fc9ffe78682e44a1a0bfa3d068725f14 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1379b38faccf6e257ae7f4603634e0550a7a9536","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-01 04:22:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48e28621c6ad0cfed4128da47cea1c25ee94f589","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-09 08:56:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf906f5d06e9e0f07f58e39e7e05812f77fa146f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-09 08:56:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_37710\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"}],"_revision_number":3},{"id":"d07cef9ed224e0e77f483c28b8aea7276a404946","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-09 08:58:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e67d60029ae3557f1847dd354121b33622c64916","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-09 08:59:05.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"82f9b4811324250c87abdaf667401600ae09fbb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-09 09:02:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0589410e1ece4c5ea9a214470cbe013d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca19d53bf045414aaef4a1b727e70d6c : SUCCESS in 5m 29s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/80d622d35b03438e92e841938f820ecb : SUCCESS in 3m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/243c09893ffe40bd9ca38c05c338182b : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":3},{"id":"b3406b5ce043e39e71950764b3890fc5dd0ae616","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-10 19:01:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ba850a749b2fc0852318c46af5f34e709f98a1c4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-10 19:02:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9466353a1bd3e40991a6c4d119d917be3adfe778","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"date":"2025-03-18 10:46:47.000000000","message":"Abandoned\n\nWill wait for Flamingo release.","accounts_in_message":[],"_revision_number":3},{"id":"de50d9230a270008d756603006fd6e83dfa58843","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-03-18 11:42:07.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c7c8aad7046fa5362ec94ee4f52297bad7cb8302","revisions":{"d815167962c527bc78abd4ff19914b2e65671d52":{"kind":"REWORK","_number":1,"created":"2025-02-27 15:58:09.000000000","uploader":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"ref":"refs/changes/18/942918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/942918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/942918/1"}}},"commit":{"parents":[{"commit":"d7f5896552eaebf4296daf3c93747846f2e919fb","subject":"Release taskflow 5.11.1 for epoxy series","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7f5896552eaebf4296daf3c93747846f2e919fb"}]}],"author":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 15:57:45.000000000","tz":0},"committer":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 15:57:45.000000000","tz":0},"subject":"Release openstackdocstheme 3.4.2","message":"Release openstackdocstheme 3.4.2\n\nChange-Id: I5505bf939cbe464d2e89908bf648c5bc8dda7b8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d815167962c527bc78abd4ff19914b2e65671d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d815167962c527bc78abd4ff19914b2e65671d52"}]},"branch":"refs/heads/master"},"26a3d04c03322f4f56ce565e85420a5798ff49d9":{"kind":"REWORK","_number":2,"created":"2025-02-27 16:40:04.000000000","uploader":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"ref":"refs/changes/18/942918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/942918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/942918/2"}}},"commit":{"parents":[{"commit":"d7f5896552eaebf4296daf3c93747846f2e919fb","subject":"Release taskflow 5.11.1 for epoxy series","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7f5896552eaebf4296daf3c93747846f2e919fb"}]}],"author":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 15:57:45.000000000","tz":0},"committer":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 16:39:21.000000000","tz":0},"subject":"Release openstackdocstheme 3.5.0","message":"Release openstackdocstheme 3.5.0\n\nChange-Id: I5505bf939cbe464d2e89908bf648c5bc8dda7b8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26a3d04c03322f4f56ce565e85420a5798ff49d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26a3d04c03322f4f56ce565e85420a5798ff49d9"}]},"branch":"refs/heads/master"},"c7c8aad7046fa5362ec94ee4f52297bad7cb8302":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-03-09 08:56:43.000000000","uploader":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/942918/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/18/942918/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/18/942918/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/18/942918/3"}}},"commit":{"parents":[{"commit":"e802ddf7bfd5b2fa24cff61dbec262908064a7fd","subject":"Merge \"[doc] Add a missing line break\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e802ddf7bfd5b2fa24cff61dbec262908064a7fd"}]}],"author":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 15:57:45.000000000","tz":0},"committer":{"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","date":"2025-03-09 08:56:43.000000000","tz":0},"subject":"Release openstackdocstheme 3.5.0","message":"Release openstackdocstheme 3.5.0\n\nChange-Id: I5505bf939cbe464d2e89908bf648c5bc8dda7b8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7c8aad7046fa5362ec94ee4f52297bad7cb8302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7c8aad7046fa5362ec94ee4f52297bad7cb8302"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
