)]}'
{"id":"openstack%2Freleases~942929","triplet_id":"openstack%2Freleases~master~I0ecaadb876e15983b5e3c3c97fa709efd0d65eb8","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-09 08:58:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-03-12 12:52:06.000000000","reason":"Change was abandoned"},"37710":{"account":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"last_update":"2025-03-12 12:52:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0ecaadb876e15983b5e3c3c97fa709efd0d65eb8","subject":"Release os-api-ref 3.0.2","status":"ABANDONED","created":"2025-02-27 17:05:57.000000000","updated":"2025-03-12 12:52:06.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3cc9293d815d8f6b471476be372d1405c216d0b7","_number":942929,"virtual_id_number":942929,"owner":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-09 09:10:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":-2,"date":"2025-03-10 19:05:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-03-09 08:58:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-03-09 08:58:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-27 17:07:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-27 17:42:45.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2025-02-27 19:06:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-08 11:44:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-03-10 19:05:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"4c427e1978afa56668f9e6e8af8f023456a51931","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"date":"2025-02-27 17:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42dee650c884040d61b4fea72e8a2c81bf67ada0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 19:06:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2548c2c453c479b9cbedfc986aea2a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84dbe3b10d74cf489ce75dcf08cb265 : SUCCESS in 6m 52s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/daf914629e8b47199aa1b660c1843930 : SUCCESS in 8m 39s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2d43b67c4b5644f9a20971c060ea75fb : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e2b7e9cdcfadf96aacff17ec2d5631def29e7e76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"real_author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-09 08:57:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_37710\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"}],"_revision_number":2},{"id":"7619d842245e17230e1ad397758f5083d3fac3f9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-09 08:58:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"41aa111b624c25b46fd0924b1602f9c878ef6e85","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-09 08:58:24.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"090150638afdeaf6b39f25efb5fb66cf2f3e9f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-09 09:10:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/007e58365d5348fdbe3e223394f14f4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5c2447a8394f37930831f9417df611 : SUCCESS in 6m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/be528016ba7b4eb4a6633d6538d05d8b : SUCCESS in 5m 13s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b0dae93a694c47ef9fd0652c9d15e809 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e820bcb14388d27d63dfe0536b66fa4b5c708e7b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-10 19:05:19.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3cc9293d815d8f6b471476be372d1405c216d0b7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"date":"2025-03-12 12:52:06.000000000","message":"Abandoned\n\nIt is a duplicate of:\n\nhttps://review.opendev.org/c/openstack/releases/+/942922","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1b3cd8bac7e29ae2b780825a975d97d589f30eb2","revisions":{"4dff27743e94b114e9faf5e6614269dfb284dfb5":{"kind":"REWORK","_number":1,"created":"2025-02-27 17:05:57.000000000","uploader":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"ref":"refs/changes/29/942929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/942929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/942929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/942929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/942929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/942929/1"}}},"commit":{"parents":[{"commit":"d2936fa3dd561d53ca940840f9f0830df489e177","subject":"Release oslotest 5.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d2936fa3dd561d53ca940840f9f0830df489e177"}]}],"author":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 17:05:47.000000000","tz":0},"committer":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 17:05:47.000000000","tz":0},"subject":"Release os-api-ref 3.0.2","message":"Release os-api-ref 3.0.2\n\nChange-Id: I0ecaadb876e15983b5e3c3c97fa709efd0d65eb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4dff27743e94b114e9faf5e6614269dfb284dfb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4dff27743e94b114e9faf5e6614269dfb284dfb5"}]},"branch":"refs/heads/master"},"1b3cd8bac7e29ae2b780825a975d97d589f30eb2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-09 08:57:35.000000000","uploader":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"real_uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/29/942929/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/29/942929/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/942929/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/942929/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/29/942929/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/29/942929/2"}}},"commit":{"parents":[{"commit":"e802ddf7bfd5b2fa24cff61dbec262908064a7fd","subject":"Merge \"[doc] Add a missing line break\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e802ddf7bfd5b2fa24cff61dbec262908064a7fd"}]}],"author":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-02-27 17:05:47.000000000","tz":0},"committer":{"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","date":"2025-03-09 08:57:35.000000000","tz":0},"subject":"Release os-api-ref 3.0.2","message":"Release os-api-ref 3.0.2\n\nChange-Id: I0ecaadb876e15983b5e3c3c97fa709efd0d65eb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b3cd8bac7e29ae2b780825a975d97d589f30eb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1b3cd8bac7e29ae2b780825a975d97d589f30eb2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
