)]}'
{"id":"openstack%2Freleases~943149","triplet_id":"openstack%2Freleases~master~I12d44a1d30e9bebee7d8e3b188b97f9cad40ceaa","project":"openstack/releases","branch":"master","topic":"epoxy-stable-branches","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-07 14:36:07.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-03-07 14:36:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I12d44a1d30e9bebee7d8e3b188b97f9cad40ceaa","subject":"[Telemetry] Create 2025.1 branch for client and non-client libs","status":"MERGED","created":"2025-03-03 10:12:25.000000000","updated":"2025-03-07 14:36:07.000000000","submitted":"2025-03-07 14:36:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943149-epoxy-stable-branches","meta_rev_id":"5de0186c18031480b7ae8366cc27a320229c7909","_number":943149,"virtual_id_number":943149,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-07 14:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-03-07 14:13:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-03-07 08:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-03-07 14:13:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-03-07 08:16:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-03 10:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-03 10:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-03 11:41:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-03-07 14:13:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"7028be518b405c75647fff166f70d677e93171e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-03 10:12:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8040d7c93107cc940b83aaf2628509abea52420f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 10:27:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/998c72986076421d8173a1fba87c5424\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe65b1ad4d14fd2bd555a077ce0b245 : SUCCESS in 7m 23s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/92366291f0c8417ab76ba778a1fcecd3 : SUCCESS in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/bf304e964dea495aa8c376927878ce57 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1840f497c39a427bc8861668605967ebc8cdbbe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-04 13:00:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"bc109bdde77d34eec6f3b1dcbf1f6802388a0bbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 13:10:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d15740cc8c4a41e3bc6dd658007f6c2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a15796220a4a10a29bbf06623b77ba : SUCCESS in 5m 23s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/cf10ba7209984f3aa7fc577fdf44a5ad : SUCCESS in 7m 30s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3509cafa5c1b436195642087ecc0f2c0 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2d2e5417b3ee4556e4914a8af9cbf53632f6ee0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-04 14:03:51.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d447faa68f0c97d129af7c869517c02969777213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 14:13:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbf9d50c0cd48189426147bde10bf0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6678b857baca4438a614099d743d3622 : SUCCESS in 7m 28s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/f8693c88907647c6a795d11545e1ef17 : SUCCESS in 4m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d3d29dfd0a644259868d75f2217e3b00 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":3},{"id":"583cc15fd11771b4d0da66c1b313d957ba6454b0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-07 08:16:23.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"31fe5f0ea475adef41d930774beb62409e710a70","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-03-07 14:13:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"32219e72dfc66c9f29ef594e6e5b5ac3a0c9bca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:14:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ee3e61058521c84f68d48b38b48a04e49f495732","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:36:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40cbf9efff5b41638bea46a470ef882c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9551988ff10f40518898700da58d686a : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b509b8896de432eadd2f72b9bd6f359 : SUCCESS in 8m 02s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/91dbb78fefd243a5b76080bae1068a4a : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":3},{"id":"5de0186c18031480b7ae8366cc27a320229c7909","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:36:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87f8b0c6f1f1b1e688edb9cf2355ba022ca5f970","revisions":{"f636ccbfdceb3e7c45c9bdf19ce64f65ca5c4117":{"kind":"REWORK","_number":1,"created":"2025-03-03 10:12:25.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/943149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/49/943149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/49/943149/1"}}},"commit":{"parents":[{"commit":"553e09ac81424a8306c663d82777ab649bd39bb1","subject":"Merge \"Release final python-barbicanclient for 2025.1 Epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/553e09ac81424a8306c663d82777ab649bd39bb1"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:54.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:54.000000000","tz":60},"subject":"[Telemetry] Create 2025.1 branch for client and non-client libs","message":"[Telemetry] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nIf the team wants to wait for a specific patch to make it to the\nlibrary (for example if the gate of the deliverable is broken and\nneeds to be fixed before branching), then someone from the team can\n-1 the patch to have it held, or update that patch with a different\ncommit SHA. If there is nothing to wait for, then please do respond\non this patch with +1.\n\nIn any case, if the patch got no response, then this will be merged\nafter Thursday, March 6th.\n\nChange-Id: I12d44a1d30e9bebee7d8e3b188b97f9cad40ceaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f636ccbfdceb3e7c45c9bdf19ce64f65ca5c4117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f636ccbfdceb3e7c45c9bdf19ce64f65ca5c4117"}]},"branch":"refs/heads/master"},"b0f99fe192c9cb58b945d97381a21962adf9b5e4":{"kind":"REWORK","_number":2,"created":"2025-03-04 13:00:10.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/943149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/49/943149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/49/943149/2"}}},"commit":{"parents":[{"commit":"089dffb7c0cd1eb90c46be3a4382c676d2662ff7","subject":"Release final python-aodhclient for 2025.1 Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/089dffb7c0cd1eb90c46be3a4382c676d2662ff7"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:54.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-04 12:59:52.000000000","tz":60},"subject":"[Telemetry] Create 2025.1 branch for client and non-client libs","message":"[Telemetry] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nIf the team wants to wait for a specific patch to make it to the\nlibrary (for example if the gate of the deliverable is broken and\nneeds to be fixed before branching), then someone from the team can\n-1 the patch to have it held, or update that patch with a different\ncommit SHA. If there is nothing to wait for, then please do respond\non this patch with +1.\n\nIn any case, if the patch got no response, then this will be merged\nafter Thursday, March 6th.\n\nChange-Id: I12d44a1d30e9bebee7d8e3b188b97f9cad40ceaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0f99fe192c9cb58b945d97381a21962adf9b5e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b0f99fe192c9cb58b945d97381a21962adf9b5e4"}]},"branch":"refs/heads/master"},"87f8b0c6f1f1b1e688edb9cf2355ba022ca5f970":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-03-04 14:03:51.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/49/943149/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/49/943149/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/49/943149/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/49/943149/3"}}},"commit":{"parents":[{"commit":"9efa5de4f59ca684e194a4d8a6ae838a962cd004","subject":"Release final python-aodhclient for 2025.1 Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9efa5de4f59ca684e194a4d8a6ae838a962cd004"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:54.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-04 14:03:37.000000000","tz":60},"subject":"[Telemetry] Create 2025.1 branch for client and non-client libs","message":"[Telemetry] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nIf the team wants to wait for a specific patch to make it to the\nlibrary (for example if the gate of the deliverable is broken and\nneeds to be fixed before branching), then someone from the team can\n-1 the patch to have it held, or update that patch with a different\ncommit SHA. If there is nothing to wait for, then please do respond\non this patch with +1.\n\nIn any case, if the patch got no response, then this will be merged\nafter Thursday, March 6th.\n\nChange-Id: I12d44a1d30e9bebee7d8e3b188b97f9cad40ceaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87f8b0c6f1f1b1e688edb9cf2355ba022ca5f970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87f8b0c6f1f1b1e688edb9cf2355ba022ca5f970"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
