)]}'
{"id":"openstack%2Freleases~943161","triplet_id":"openstack%2Freleases~master~Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1","project":"openstack/releases","branch":"master","topic":"epoxy-stable-branches","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2025-03-06 09:16:32.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2025-03-06 09:16:32.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-06 09:16:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1","subject":"[manila] Create 2025.1 branch for client and non-client libs","status":"MERGED","created":"2025-03-03 10:14:31.000000000","updated":"2025-03-06 09:16:32.000000000","submitted":"2025-03-06 09:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943161-epoxy-stable-branches","meta_rev_id":"9ee3e00697825ad0bf87a44d0e6ee2c73d1d3725","_number":943161,"virtual_id_number":943161,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-06 09:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2025-03-06 08:44:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-03-06 07:52:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2025-03-06 08:44:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-03-05 22:54:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-03 10:15:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-03 10:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-03 11:41:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2025-03-03 11:41:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2025-03-06 08:44:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"6033bb62e08adb73bcc8a40cc28ec9a70a78a952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-03 10:14:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b29263045291a1961b52ae203e1ef80afe29b1b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 10:30:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85ec27417fad444ab4d744d94498adce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d92ee4eb30f48429666882d2db5d8fb : SUCCESS in 7m 25s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e3887cd81f754c15b601200218710d6c : SUCCESS in 5m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7338d4168947499a8752cfa70054fe77 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f356614f3ab444e7ff4374a2db064dda144ce0f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-05 10:57:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b9b25ab9b3847d0f223cb8481ca41a66d9a2f2bd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-05 10:58:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f32eebaae85928eaff014f94c52d639d7f4879d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-05 10:58:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"15786252b5c58d939a8fe0ff2f3ad65cc52b0c64","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 10:58:36.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"f01f6655f193a465c5ae5e51c36b95b04a4c2d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 11:19:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6c5600e320545589b40ff9c84240791\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b88fccda1c549a1956511847b893d5e : SUCCESS in 6m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/da6d4b4422fa437da77f8f9d2b7eaa58 : FAILURE in 4m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/42f88b824b8f4edc9a2e3753c4fe7b7e : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ae63565322d12c1208bd22d4ec2051fb22ea1c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 12:09:57.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6c2569a2d73a7de16911928bc40bdd6cf377df61","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 12:10:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"821f605f7aa30a901a0169d85648b8ee0680c214","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-05 12:11:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"53676de0d4a02a65316ef069bc60f7f073df4e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 12:22:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48089b7cce2b4c7c89f8ca380ee4f213\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04ac881b25c473fbb5254f4e0409ab8 : SUCCESS in 6m 51s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6954f67c21fb40f88b0bbf6e6d5248ac : SUCCESS in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8d0b1f4d82ae41c58e3a8b75ecd41d55 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"9e873eebee87fcb3756d92e77542139de0f6f675","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2025-03-05 12:36:29.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"832ed56ae0d5627dc6ffd5a816eb6bdd3e5f0296","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 12:37:32.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"0e4eab195755992b11e59b3bdb8c424da89c4e9d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-05 15:17:21.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09c617aff8b7c7fcd7772e488ef858070d2ad28c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-05 17:03:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ec0a222fe0a163f40c9065116fbc0072ce9d5442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-05 17:06:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7a0774c5e0b67a642d603691eb07eaec62c55112","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-05 17:07:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"12fa08699d171736eaf28f087900203e38f8d1be","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 17:07:09.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"ecfb86e9806f2e80ad39031c105db78f925f1945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 17:16:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d54df7391924256af2f2123759005de\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef02f83b2ad42e2876e784d3cbd428f : SUCCESS in 4m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6857b6ba2ec44ff7ad677fc13d2b7641 : SUCCESS in 3m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/55ac523a3b074980af6c7b90a5aff876 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":4},{"id":"031c317931e03ca6d02a5112febf387bac910577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-05 22:52:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d8ccfad853dd9a97cdd1f0ae3b489f8cde57a634","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:53:44.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"722ee2307f9b5e8d6ee99c54322ae39f8adf11d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2025-03-05 22:53:55.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":6},{"id":"8a23f4d1545afbb0c45fab70fe1b115c3f5e490f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 22:54:38.000000000","message":"Patch Set 6: PTL-Approved+1","accounts_in_message":[],"_revision_number":6},{"id":"cba0fb0ef42351b87ec8902cb8d00945bdeef318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 23:06:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69ecd8bc80149f8ba944f3f5adeef30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d1b7a4caa045cc912fb098a8a33ce5 : SUCCESS in 5m 19s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/68d1d94b3038412c95cdb66d182d36de : SUCCESS in 6m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6f74ca7e606747319ad69febdd52986f : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":6},{"id":"5793b138b01993e33ef77ca7b463a709567d4686","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-06 07:52:43.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c43bfe6d1af81824426d13fdef11bc2da985e1c0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-03-06 08:44:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a057f1960d82053d05087c7549e9aa54a24964ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 08:45:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"624d965523f4e49bf463f48c069067f5f258045a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:16:31.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b829c3a078f4cf3a795fc5aa65f4a5d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e27404fb8a4c6385e5adb9ddee60d9 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf7d706f6584624ab878ee6b3bb064f : SUCCESS in 7m 59s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6b33c8975af84b05841ac28668477949 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":6},{"id":"9ee3e00697825ad0bf87a44d0e6ee2c73d1d3725","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-06 09:16:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eb0eadcd4171991df41b296a7a02b3047ef19f4d","revisions":{"634a08a1ca1be42e47d4acb7925dddd91409b929":{"kind":"REWORK","_number":1,"created":"2025-03-03 10:14:31.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/61/943161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/943161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/943161/1"}}},"commit":{"parents":[{"commit":"553e09ac81424a8306c663d82777ab649bd39bb1","subject":"Merge \"Release final python-barbicanclient for 2025.1 Epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/553e09ac81424a8306c663d82777ab649bd39bb1"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:56.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:56.000000000","tz":60},"subject":"[manila] Create 2025.1 branch for client and non-client libs","message":"[manila] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nIf the team wants to wait for a specific patch to make it to the\nlibrary (for example if the gate of the deliverable is broken and\nneeds to be fixed before branching), then someone from the team can\n-1 the patch to have it held, or update that patch with a different\ncommit SHA. If there is nothing to wait for, then please do respond\non this patch with +1.\n\nIn any case, if the patch got no response, then this will be merged\nafter Thursday, March 6th.\n\nChange-Id: Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634a08a1ca1be42e47d4acb7925dddd91409b929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/634a08a1ca1be42e47d4acb7925dddd91409b929"}]},"branch":"refs/heads/master"},"c7e001062bd078351209aeaf8587c9c1b40bf779":{"kind":"REWORK","_number":2,"created":"2025-03-05 10:57:41.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/61/943161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/943161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/943161/2"}}},"commit":{"parents":[{"commit":"26e39de74045876d596791e16608be920909e86b","subject":"Merge \"Release final python-cinderclient for 2025.1 Epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26e39de74045876d596791e16608be920909e86b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:56.000000000","tz":60},"committer":{"name":"Carlos da Silva","email":"ces.eduardo98@gmail.com","date":"2025-03-05 10:57:40.000000000","tz":-180},"subject":"[manila] Create 2025.1 branch for client and non-client libs","message":"[manila] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nIf the team wants to wait for a specific patch to make it to the\nlibrary (for example if the gate of the deliverable is broken and\nneeds to be fixed before branching), then someone from the team can\n-1 the patch to have it held, or update that patch with a different\ncommit SHA. If there is nothing to wait for, then please do respond\non this patch with +1.\n\nIn any case, if the patch got no response, then this will be merged\nafter Thursday, March 6th.\n\nChange-Id: Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7e001062bd078351209aeaf8587c9c1b40bf779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c7e001062bd078351209aeaf8587c9c1b40bf779"}]},"branch":"refs/heads/master"},"b84e9c724c88e128531c37f6f01ed4a6884b1f2b":{"kind":"REWORK","_number":3,"created":"2025-03-05 12:09:57.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/61/943161/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/943161/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/943161/3"}}},"commit":{"parents":[{"commit":"26e39de74045876d596791e16608be920909e86b","subject":"Merge \"Release final python-cinderclient for 2025.1 Epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/26e39de74045876d596791e16608be920909e86b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:56.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-03-05 12:09:57.000000000","tz":0},"subject":"[manila] Create 2025.1 branch for client and non-client libs","message":"[manila] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nIf the team wants to wait for a specific patch to make it to the\nlibrary (for example if the gate of the deliverable is broken and\nneeds to be fixed before branching), then someone from the team can\n-1 the patch to have it held, or update that patch with a different\ncommit SHA. If there is nothing to wait for, then please do respond\non this patch with +1.\n\nIn any case, if the patch got no response, then this will be merged\nafter Thursday, March 6th.\n\nChange-Id: Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b84e9c724c88e128531c37f6f01ed4a6884b1f2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b84e9c724c88e128531c37f6f01ed4a6884b1f2b"}]},"branch":"refs/heads/master"},"5db167a1aabeff6aadaa31be7a16b85a1659858f":{"kind":"REWORK","_number":4,"created":"2025-03-05 17:06:18.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/943161/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/943161/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/943161/4"}}},"commit":{"parents":[{"commit":"a433876bab973be9e097beaabe395473f220a6c0","subject":"Bugfix release for epoxy manilaclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a433876bab973be9e097beaabe395473f220a6c0"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:56.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-03-05 17:05:49.000000000","tz":-480},"subject":"[manila] Create 2025.1 branch for client and non-client libs","message":"[manila] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nChange-Id: Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1\n(cherry picked from commit b84e9c724c88e128531c37f6f01ed4a6884b1f2b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5db167a1aabeff6aadaa31be7a16b85a1659858f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5db167a1aabeff6aadaa31be7a16b85a1659858f"}]},"branch":"refs/heads/master"},"1ffd6b32d4600dc4d620429f2acee2c5fea64ecf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-03-05 22:52:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/943161/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/943161/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/943161/5"}}},"commit":{"parents":[{"commit":"23b136b2060d01d1d1ad3a724eaea93e48909cb4","subject":"Bugfix release for epoxy manilaclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23b136b2060d01d1d1ad3a724eaea93e48909cb4"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:56.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-03-05 22:52:54.000000000","tz":0},"subject":"[manila] Create 2025.1 branch for client and non-client libs","message":"[manila] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nChange-Id: Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1\n(cherry picked from commit b84e9c724c88e128531c37f6f01ed4a6884b1f2b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1ffd6b32d4600dc4d620429f2acee2c5fea64ecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1ffd6b32d4600dc4d620429f2acee2c5fea64ecf"}]},"branch":"refs/heads/master","description":"Rebase"},"eb0eadcd4171991df41b296a7a02b3047ef19f4d":{"kind":"REWORK","_number":6,"created":"2025-03-05 22:53:55.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/61/943161/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/943161/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/943161/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/943161/6"}}},"commit":{"parents":[{"commit":"23b136b2060d01d1d1ad3a724eaea93e48909cb4","subject":"Bugfix release for epoxy manilaclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23b136b2060d01d1d1ad3a724eaea93e48909cb4"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-03 10:09:56.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2025-03-05 22:53:54.000000000","tz":0},"subject":"[manila] Create 2025.1 branch for client and non-client libs","message":"[manila] Create 2025.1 branch for client and non-client libs\n\nThis creates the stable/2025.1 branch for all client and non-client\nlibraries now that we are past the library freeze.\n\nChange-Id: Icd7141ece5f12467d0bbcbb1dfea209cc79b5cc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb0eadcd4171991df41b296a7a02b3047ef19f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb0eadcd4171991df41b296a7a02b3047ef19f4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
