)]}'
{"id":"openstack%2Freleases~943975","triplet_id":"openstack%2Freleases~master~Id9ea9223266b2796f0d5c7e3daddd86c085e4859","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"37710":{"account":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"last_update":"2025-08-22 14:26:47.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id9ea9223266b2796f0d5c7e3daddd86c085e4859","subject":"Tool update proposal - list_unreleased_changes. The tool will now return branches with pending changes while omitting the empty ones.","status":"ABANDONED","created":"2025-03-10 17:20:30.000000000","updated":"2025-08-22 14:26:47.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"d49af10b9fba711a19182189720a7e9fb3367a5a","_number":943975,"virtual_id_number":943975,"owner":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-10 17:32:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-10 17:32:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-10 18:11:15.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"ffba4723b4d337de6d5e524180b5a2149791d60d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"date":"2025-03-10 17:20:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc5e5ee86a5be0d3264663b731c47c56b41365f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 17:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74eb208138974226accb8684fa82dc82\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7ccdcbf3304c39a8ce906bd99a6a9d : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68c36d30fbfe45ddaaa0c070752b772a : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/152a35d3042e492b9b3ec73a61f01a6b : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9aff0ecbec80401f841fb63bf507b379 : SUCCESS in 3m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce4cf202d3e247ed9571a675711c71e2 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8f18fc34cba4ed4be47b5035f005bf3 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8766b4c4e549df885bb8a474b5ee47 : SUCCESS in 6m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c2da83b8976f4a13af489f7b3801506b : SUCCESS in 2m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4de8698a19a84a9ca5b9567c41ac1383 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1039bc6a1160ab7c01f717a0ca268864225d5712","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-10 18:11:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d49af10b9fba711a19182189720a7e9fb3367a5a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-08-22 14:26:47.000000000","message":"Abandoned\n\nNo more activity on the patch, please restore it if you plan to work on it again","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75aedf553451ae88e7b6388c261f34e66b14fb53","revisions":{"75aedf553451ae88e7b6388c261f34e66b14fb53":{"kind":"REWORK","_number":1,"created":"2025-03-10 17:20:30.000000000","uploader":{"_account_id":37710,"name":"Kacper Kulakowski","email":"kkulakow@redhat.com","username":"Kacperrh"},"ref":"refs/changes/75/943975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/943975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/943975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/943975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/943975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/943975/1"}}},"commit":{"parents":[{"commit":"5b69546d6865243f931fd5638d0ac055b30e9e13","subject":"Merge \"Release taskflow 5.12.0 for epoxy series\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5b69546d6865243f931fd5638d0ac055b30e9e13"}]}],"author":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-03-10 17:20:02.000000000","tz":0},"committer":{"name":"Kacper","email":"kkulakow@redhat.com","date":"2025-03-10 17:20:02.000000000","tz":0},"subject":"Tool update proposal - list_unreleased_changes. The tool will now return branches with pending changes while omitting the empty ones.","message":"Tool update proposal - list_unreleased_changes. The tool will now return branches with pending changes while omitting the empty ones.\n\nChange-Id: Id9ea9223266b2796f0d5c7e3daddd86c085e4859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75aedf553451ae88e7b6388c261f34e66b14fb53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/75aedf553451ae88e7b6388c261f34e66b14fb53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:PTL-Approved\u003e0"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
