)]}'
{"id":"openstack%2Freleases~944774","triplet_id":"openstack%2Freleases~master~I5a42c8936e5942187813ef2bfa7f2660431c001f","project":"openstack/releases","branch":"master","topic":"horizon","attention_set":{},"removed_from_attention_set":{"29313":{"account":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"last_update":"2025-03-19 16:26:24.000000000","reason":"Change was submitted"},"6914":{"account":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"last_update":"2025-03-19 16:26:24.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-03-19 16:19:23.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-19 16:26:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a42c8936e5942187813ef2bfa7f2660431c001f","subject":"[horizon] Cut stable/2025.1 branch","status":"MERGED","created":"2025-03-17 13:06:20.000000000","updated":"2025-03-19 16:26:24.000000000","submitted":"2025-03-19 16:26:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944774-horizon","meta_rev_id":"49900a4bf2df0ab63dd58d6164f8edd29e19bc1d","_number":944774,"virtual_id_number":944774,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-19 16:26:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":2,"date":"2025-03-19 16:19:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-19 15:09:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":1,"date":"2025-03-19 16:19:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-03-19 14:32:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}],"reviewers":{"REVIEWER":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-17 13:07:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-17 13:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-17 13:32:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"REVIEWER"},{"updated":"2025-03-17 13:32:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2025-03-18 17:18:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-03-18 17:51:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"67cad28ba913ead1892290153a86a419de7cf0f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-17 13:06:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea401b95c3ba01f6f2a76fa59afceaa09ed73c3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 13:11:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4552e67a894455ab68528f94c558ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5f14c00850947d384e7fb3c46077d6b : SUCCESS in 4m 27s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/75baf805474c4d21a0dd8fbb4279fb95 : SUCCESS in 2m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c673388dc88c489bb7c2a9f2a0423c90 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":1},{"id":"30ce25628ed059cfe15999d9112b7ca6fc294edd","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-03-18 16:57:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"85d41912d1c589b82ffedafafbc835508b1a1b99","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 16:58:13.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"6018096de9d4245f33277bf2830a1fb3d6ba50df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-03-18 17:16:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"65e55a2498097f55b39eb54f361f8d8d5949dfca","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 17:16:52.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a8ee29e1841f34cb25ec5d9130b979689fc3a2c4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-18 17:18:41.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8acffbd332c7dcdf30d4e1af14586eab6a935db3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-03-18 17:21:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":3},{"id":"093729d66c1385bd56b1490f393ade3d71ce088b","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-03-18 17:22:02.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3d5ed6ab503b9612465a4ac8be11b0d1447e2f7b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 17:25:16.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"a0a7e13baf11ebb7bfa78332d0da47ce91bd17de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 17:35:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25cc34c381de4201a3e6e4cba85a8afa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5de106b6c154b3fadbc869a8cda84cc : SUCCESS in 8m 00s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/8b31b421e2354cf08fe5472c425eea5b : SUCCESS in 7m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6b0371b4e78e499a81e27eb55bd03beb : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"80ba1aa572c3ecde49f958266c71cfa193a6bf0b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-18 17:51:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"be96da7968e8ed0be75d1d792c51ef66c03be289","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-19 09:31:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a713a46e005d6c4c55932af395d7ec100baadd4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-03-19 14:31:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"dbfdbb8c39f30b2469173911f0738f34307ee091","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 14:32:08.000000000","message":"Patch Set 4: PTL-Approved+1","accounts_in_message":[],"_revision_number":4},{"id":"48052f82656ca94978c342cde9d9b4292c35c0d0","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2025-03-19 14:32:17.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b3fcb294a139a5d205b37846d1de03b24391e90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 14:42:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c1746ab36443698067bb369272110b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b27ac8bd1994527871afbe71a4f23e1 : SUCCESS in 6m 01s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/23a6d775a0744cb797437c95493fb100 : SUCCESS in 2m 41s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6628419571004c2a832755e07199c224 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":4},{"id":"3e69e85f8213a983dad0a9694119976363818166","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-19 15:09:44.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1fa82e7b2f62033b0b1d59be208f1993cb34e2dc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-19 16:19:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f7e4752bfe88012ce3478d1983899c82f0b06526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 16:19:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"66070c50434104c8c05865ad0a96b86c20ce8920","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 16:26:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b16a114354654e1eaba0c971c388e30d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af461528d3ea4f028f2b4315e79906d4 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9544dc2a2cf484b9be8ff626e020681 : SUCCESS in 6m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/9691b8b6eacc455db70b9800237317df : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":4},{"id":"49900a4bf2df0ab63dd58d6164f8edd29e19bc1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 16:26:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"899c631a82b6ad4b0ada657efe25077c76e7f53f","revisions":{"eb9e4f5ca3b3ed4b22eb4778685db37006ee574d":{"kind":"REWORK","_number":1,"created":"2025-03-17 13:06:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/74/944774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/944774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/944774/1"}}},"commit":{"parents":[{"commit":"67d0aef9b530399638818532c1444644c48fa417","subject":"Merge \"Release magnum-ui RC1 for 2025.1 Epoxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67d0aef9b530399638818532c1444644c48fa417"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-17 13:05:33.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-17 13:05:33.000000000","tz":60},"subject":"[horizon] Cut stable/2025.1 branch","message":"[horizon] Cut stable/2025.1 branch\n\nThis patch cuts missing stable/2025.1 branches for deliverables that are\nmissing it.\n\nWe are in Feature Freeze period already, but if the team needs to do a\nanother final release because of any release critical fixes, then please\nindicate it with a -1.\nIf there is nothing to wait for, then please do respond on this patch\nwith +1.\n\nIn any case, if the patch gets no response, then this will be merged\nafter Thursday, March 20th.\n\nChange-Id: I5a42c8936e5942187813ef2bfa7f2660431c001f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9e4f5ca3b3ed4b22eb4778685db37006ee574d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9e4f5ca3b3ed4b22eb4778685db37006ee574d"}]},"branch":"refs/heads/master"},"87fb629b74529458d1171ee7a99992a0c46547da":{"kind":"REWORK","_number":2,"created":"2025-03-18 17:16:02.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/74/944774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/944774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/944774/2"}}},"commit":{"parents":[{"commit":"efed3c3bd04b3f2cd2b8ed913e3567aa6183a0ff","subject":"Release nova RC1 for 2025.1 Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/efed3c3bd04b3f2cd2b8ed913e3567aa6183a0ff"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-17 13:05:33.000000000","tz":60},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2025-03-18 17:15:55.000000000","tz":-300},"subject":"[horizon] Cut stable/2025.1 branch","message":"[horizon] Cut stable/2025.1 branch\n\nThis patch cuts missing stable/2025.1 branches for deliverables that are\nmissing it.\n\nWe are in Feature Freeze period already, but if the team needs to do a\nanother final release because of any release critical fixes, then please\nindicate it with a -1.\nIf there is nothing to wait for, then please do respond on this patch\nwith +1.\n\nIn any case, if the patch gets no response, then this will be merged\nafter Thursday, March 20th.\n\nChange-Id: I5a42c8936e5942187813ef2bfa7f2660431c001f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87fb629b74529458d1171ee7a99992a0c46547da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/87fb629b74529458d1171ee7a99992a0c46547da"}]},"branch":"refs/heads/master"},"dd0fddf31192931c3030993407836c2a4b54c28a":{"kind":"REWORK","_number":3,"created":"2025-03-18 17:21:57.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/74/944774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/944774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/944774/3"}}},"commit":{"parents":[{"commit":"efed3c3bd04b3f2cd2b8ed913e3567aa6183a0ff","subject":"Release nova RC1 for 2025.1 Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/efed3c3bd04b3f2cd2b8ed913e3567aa6183a0ff"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-17 13:05:33.000000000","tz":60},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2025-03-18 17:21:48.000000000","tz":-300},"subject":"[horizon] Cut stable/2025.1 branch","message":"[horizon] Cut stable/2025.1 branch\n\nThis patch cuts missing stable/2025.1 branches for deliverables that are\nmissing it.\n\nWe are in Feature Freeze period already, but if the team needs to do a\nanother final release because of any release critical fixes, then please\nindicate it with a -1.\nIf there is nothing to wait for, then please do respond on this patch\nwith +1.\n\nIn any case, if the patch gets no response, then this will be merged\nafter Thursday, March 20th.\n\nChange-Id: I5a42c8936e5942187813ef2bfa7f2660431c001f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd0fddf31192931c3030993407836c2a4b54c28a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd0fddf31192931c3030993407836c2a4b54c28a"}]},"branch":"refs/heads/master"},"899c631a82b6ad4b0ada657efe25077c76e7f53f":{"kind":"REWORK","_number":4,"created":"2025-03-19 14:31:18.000000000","uploader":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"ref":"refs/changes/74/944774/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/74/944774/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/74/944774/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/74/944774/4"}}},"commit":{"parents":[{"commit":"efed3c3bd04b3f2cd2b8ed913e3567aa6183a0ff","subject":"Release nova RC1 for 2025.1 Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/efed3c3bd04b3f2cd2b8ed913e3567aa6183a0ff"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-17 13:05:33.000000000","tz":60},"committer":{"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","date":"2025-03-19 14:30:59.000000000","tz":-300},"subject":"[horizon] Cut stable/2025.1 branch","message":"[horizon] Cut stable/2025.1 branch\n\nThis patch cuts missing stable/2025.1 branches for deliverables that are\nmissing it.\n\nWe are in Feature Freeze period already, but if the team needs to do a\nanother final release because of any release critical fixes, then please\nindicate it with a -1.\nIf there is nothing to wait for, then please do respond on this patch\nwith +1.\n\nIn any case, if the patch gets no response, then this will be merged\nafter Thursday, March 20th.\n\nChange-Id: I5a42c8936e5942187813ef2bfa7f2660431c001f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/899c631a82b6ad4b0ada657efe25077c76e7f53f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/899c631a82b6ad4b0ada657efe25077c76e7f53f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
