)]}'
{"id":"openstack%2Freleases~945285","triplet_id":"openstack%2Freleases~master~Ifdb4fb28ced8dc45f69403ff629936bbdf61c1d6","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2025-03-26 09:00:40.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2025-03-26 09:00:40.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-03-26 08:52:28.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-24 10:04:34.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"6282":{"account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"last_update":"2025-03-25 01:33:22.000000000","reason":"\u003cGERRIT_ACCOUNT_6282\u003e replied on the change","reason_account":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"}}},"hashtags":[],"change_id":"Ifdb4fb28ced8dc45f69403ff629936bbdf61c1d6","subject":"Release RC2 for Skyline-Console","status":"MERGED","created":"2025-03-24 03:26:11.000000000","updated":"2025-03-26 11:28:07.000000000","submitted":"2025-03-26 09:00:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945285","meta_rev_id":"9069f1400e02d7264a6cbb7b423b76f6cea89de1","_number":945285,"virtual_id_number":945285,"owner":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-26 09:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-26 08:52:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-24 10:04:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-03-25 01:33:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-26 08:52:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-03-24 09:13:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 03:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-24 03:40:42.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-03-24 03:40:42.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2025-03-24 10:28:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"state":"REVIEWER"},{"updated":"2025-03-26 05:33:48.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-03-26 11:23:36.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"}],"messages":[{"id":"73038d89b21dd5b2754d87d56a84cbb785113269","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-24 03:26:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88dd2480cabc1db45062e51721325b0a948fd3b8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 03:26:57.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"3a2d59e71ab184803f4321c2930ed5d0dce25e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 03:36:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a3170ba5ccb34ca0bb671bb02b2a4903\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7a6f3b5a6024158aee67ae16acaa55e : FAILURE in 5m 53s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/f583376437564c3cb10b930d5d7919fa : FAILURE in 2m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/de326185ac17466dbf71ee072a61470e : FAILURE in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1b638a9c3562656d99d49f2a735fc4ca7b717818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-24 03:39:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"553f0abdc8ef54f7f1fd64ad2ed85948cbe119d8","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 03:40:38.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"c66d2275107937d31a5c1c872b9d79e3cae03850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 03:49:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c67633413f0e4c6ebe4ffddd8c621531\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51cf6d0528334a85bc9f3a68d013cbda : SUCCESS in 4m 53s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/fd01cbc7eac040498be01dc7123c5d32 : SUCCESS in 3m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/19dcacde97e24f53b102e9159b9ad523 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ad07fdec5e7233c2bfe4881082822bde2fdbaa61","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-24 08:53:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"717713cdc36f284f60a78ae89c36aadeb31f150a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-24 09:12:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b4e3e69af515a37a594717404ff15398f5cd4dd9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 09:13:25.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"97e0604696ca40d09136e97efe4febd2c08a6f4d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-24 09:14:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b248f441acd83d7476cfe5229a91786de461716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 09:23:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bd49be1f230415c97cebaa47762bea0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877a1e38c289438faf3a92df6d356321 : SUCCESS in 6m 34s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/c77ed377eea54b4eba7c60c4ecb5ea45 : SUCCESS in 5m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5425f5d575764cb8b98bde9dcbffabfd : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":3},{"id":"47562fa5667e1239d44d372b7e75f53a9b6c58dd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-24 10:04:34.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5a51d6e881528953c61a2a70e807e0a6e8aa8b5c","author":{"_account_id":6282,"name":"Wu Wenxiang","email":"wu.wenxiang@algoblu.com","username":"wu-wenxiang"},"date":"2025-03-25 01:33:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f6ef3e18084b38699bbde6e8e207bcbec8276266","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2025-03-26 05:33:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a99fc1a4e1378a0041048366e733581fde2de04c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-26 08:52:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e54fedcc005d99090513919d3760a44f090b5af3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 08:52:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d345a9c2072fb56457009a940623b6a90f90dcf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 09:00:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d25ebde7d7424b8db35c93250fb812\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a804071eeb534a9c94fe733157c2e954 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64220252923b426495f879abb70671d9 : SUCCESS in 5m 35s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/47c3366e25f447a38a6a24baede95fa4 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"78a35ea08b7c5a86ef6a09977d701d4238e9bb62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 09:00:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5ceb055a07f0dce1b231b93316367b3c670e5875","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-26 11:23:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9069f1400e02d7264a6cbb7b423b76f6cea89de1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-03-26 11:28:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"450ffedc87507a0a2ad4fcc8db706cd078e3ac54","revisions":{"3c6dc4f8c6a6ff0c1f4bb68f7f65248f6f7eb1d2":{"kind":"REWORK","_number":1,"created":"2025-03-24 03:26:11.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/85/945285/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/945285/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/945285/1"}}},"commit":{"parents":[{"commit":"2f86d786147f9839b075c91a6630d55e46832ad8","subject":"Merge \"Publish the 2025.2/Flamingo key for future use\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f86d786147f9839b075c91a6630d55e46832ad8"}]}],"author":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2025-03-24 03:24:33.000000000","tz":480},"committer":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2025-03-24 03:24:33.000000000","tz":480},"subject":"Release RC2 for Skyline-Console","message":"Release RC2 for Skyline-Console\n\nUpdate package for release. So releasing an RC2.\n\nChange-Id: Ifdb4fb28ced8dc45f69403ff629936bbdf61c1d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c6dc4f8c6a6ff0c1f4bb68f7f65248f6f7eb1d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3c6dc4f8c6a6ff0c1f4bb68f7f65248f6f7eb1d2"}]},"branch":"refs/heads/master"},"33bb02d6608a956688480fcedaf04c2ec76e9d84":{"kind":"REWORK","_number":2,"created":"2025-03-24 03:39:55.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/85/945285/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/945285/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/945285/2"}}},"commit":{"parents":[{"commit":"2f86d786147f9839b075c91a6630d55e46832ad8","subject":"Merge \"Publish the 2025.2/Flamingo key for future use\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f86d786147f9839b075c91a6630d55e46832ad8"}]}],"author":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2025-03-24 03:24:33.000000000","tz":480},"committer":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2025-03-24 03:39:41.000000000","tz":480},"subject":"Release RC2 for Skyline-Console","message":"Release RC2 for Skyline-Console\n\nUpdate package for release. So releasing an RC2.\n\nChange-Id: Ifdb4fb28ced8dc45f69403ff629936bbdf61c1d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33bb02d6608a956688480fcedaf04c2ec76e9d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/33bb02d6608a956688480fcedaf04c2ec76e9d84"}]},"branch":"refs/heads/master"},"450ffedc87507a0a2ad4fcc8db706cd078e3ac54":{"kind":"REWORK","_number":3,"created":"2025-03-24 09:12:40.000000000","uploader":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"ref":"refs/changes/85/945285/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/945285/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/945285/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/945285/3"}}},"commit":{"parents":[{"commit":"2f86d786147f9839b075c91a6630d55e46832ad8","subject":"Merge \"Publish the 2025.2/Flamingo key for future use\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f86d786147f9839b075c91a6630d55e46832ad8"}]}],"author":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2025-03-24 03:24:33.000000000","tz":480},"committer":{"name":"zhuboxiang","email":"zhu.boxiang@99cloud.net","date":"2025-03-24 09:12:27.000000000","tz":480},"subject":"Release RC2 for Skyline-Console","message":"Release RC2 for Skyline-Console\n\nUpdate package for release. So releasing an RC2.\n\nChange-Id: Ifdb4fb28ced8dc45f69403ff629936bbdf61c1d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/450ffedc87507a0a2ad4fcc8db706cd078e3ac54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/450ffedc87507a0a2ad4fcc8db706cd078e3ac54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
