)]}'
{"id":"openstack%2Freleases~945561","triplet_id":"openstack%2Freleases~master~I560c71af123c2212653e4104afea8e91861c6e99","project":"openstack/releases","branch":"master","topic":"tools-list-rc-updates","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-03-27 13:40:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I560c71af123c2212653e4104afea8e91861c6e99","subject":"[tool] Exclude trailing projects in rc update script","status":"MERGED","created":"2025-03-26 11:30:32.000000000","updated":"2025-03-27 13:40:16.000000000","submitted":"2025-03-27 13:40:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945561-tools-list-rc-updates","meta_rev_id":"71dea8f21c5a5313baf254abfb5e18629a7597ce","_number":945561,"virtual_id_number":945561,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-27 13:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-26 12:17:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-27 13:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-27 13:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-26 12:17:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 11:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 12:17:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-03-27 13:07:40.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"364850320fcbd232128870e0abc856b7b058edc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-26 11:30:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7310fe2c2bdb98982ba5fcb82f31f3e174a1d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 11:51:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75140f7ac0324d33a15f4171aebe6400\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70e21d72dfda4243b909f0ac1cd471d6 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db5d459b597d428ba2d9f9428231f311 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/354979e9310744fb8930598f2441ea03 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a5813adc675484ea24ea07d906b32b2 : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/415fa4be80094cbe807be76fa40cb059 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a0a2c4a225c47bfb477efcfb4379bf7 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc40f46f37a54d25b0859b2dda7e2ac7 : SUCCESS in 3m 59s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/f52e70ae9fa847659d8b96d254623a28 : SUCCESS in 2m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b2d0a79372fd4b13a681913f2de0a57d : SUCCESS in 2m 23s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/eb05dc44551c4a859da8140ed63005d2 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6a75a61b327b75a23e29b77450186ae6389f6277","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-26 12:17:43.000000000","message":"Patch Set 1: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87be58dd6aadebf121215929df8887ae53e6e076","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-03-27 13:07:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"349b8e88292fa99274804d7d82900dca4858a0e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:08:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29263084b77e7420c36784d8e9674e238760c9f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:40:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0cf1d25f6ee45cc9618225754fdfcf9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd6fb06136b47e6b16ad89a60a12bac : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/496b5ffff0f54baf8642ca1d30e6be35 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de297c5971ae47ec854d59df87cf51a1 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e9da95ef65448da992675ae24c3c099 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/445a10020e624cfeadfdd9428fa29212 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43d8ec47f60e4b208b92ce688b90f8d6 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e7ee49d7b54ee39ea19eedba51b03d : SUCCESS in 7m 11s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/00f7b80ef0a04ab5ac54ba0501afa9a1 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"71dea8f21c5a5313baf254abfb5e18629a7597ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:40:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"925de9c3b20bf301820f917adf865fd113ff689d","revisions":{"925de9c3b20bf301820f917adf865fd113ff689d":{"kind":"REWORK","_number":1,"created":"2025-03-26 11:30:32.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/61/945561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/61/945561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/945561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/945561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/61/945561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/61/945561/1"}}},"commit":{"parents":[{"commit":"2f86d786147f9839b075c91a6630d55e46832ad8","subject":"Merge \"Publish the 2025.2/Flamingo key for future use\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f86d786147f9839b075c91a6630d55e46832ad8"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-26 11:20:16.000000000","tz":60},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-03-26 11:20:16.000000000","tz":60},"subject":"[tool] Exclude trailing projects in rc update script","message":"[tool] Exclude trailing projects in rc update script\n\nThe list_rc_updates.sh script is used to check whether a new final\nrelease candidate is needed for any project before doing the final\nrelease itself. Since trailing projects are doing postponed releases\n~2 months after the coordinated final release of a series, those\ndeliverables can be excluded during this check.\n\nChange-Id: I560c71af123c2212653e4104afea8e91861c6e99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/925de9c3b20bf301820f917adf865fd113ff689d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/925de9c3b20bf301820f917adf865fd113ff689d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
