)]}'
{"id":"openstack%2Freleases~947147","triplet_id":"openstack%2Freleases~master~I0634d4045bfaa12790e04aef745b8612f550af18","project":"openstack/releases","branch":"master","attention_set":{},"removed_from_attention_set":{"6469":{"account":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"last_update":"2025-04-17 09:10:00.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-04-17 09:10:00.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-04-17 09:10:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0634d4045bfaa12790e04aef745b8612f550af18","subject":"Release Octavia stable/2023.2 bugfixes","status":"MERGED","created":"2025-04-14 19:04:09.000000000","updated":"2025-04-18 17:26:33.000000000","submitted":"2025-04-17 09:10:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947147","meta_rev_id":"813481e5b311a88a2cf2ea2e349be6bd62b12762","_number":947147,"virtual_id_number":947147,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-17 09:10:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-04-17 09:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-04-16 11:33:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-04-17 09:01:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-04-14 19:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-14 19:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-15 12:26:44.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2025-04-16 11:33:22.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-04-16 19:01:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2025-04-17 09:01:11.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"dd8049aacfa8208d8d567b8997b8e0e36c64e39a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-14 19:04:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6424c24daa0a6db724f42c880d632b2228c7bc33","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 19:05:06.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"6a5c3f25a697c15a3db4b83baf152850f966e60b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-14 19:06:18.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* PTL-Approved+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE\")\n","accounts_in_message":[],"_revision_number":2},{"id":"85095fcfb6f7c7051a32d89dfbb43862427a6c73","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 19:07:05.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"abbd977c28a9b5acb3a24a83bac8e63e1aadc467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 19:13:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea4a063044341e6b97d33d671c1e952\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61769eab45a4dbc870940ad75bd4db5 : SUCCESS in 7m 17s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/05078ce5b41e438c97e1a6475686317d : SUCCESS in 6m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/aaff49b2ebf84c38a4ba2a6a9e83c8fe : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8e19463eae4c7a98982671e3a6780b56127ab0b3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-16 11:33:22.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6dfb5f7c169c2760322f331489dfeba659aefc21","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-04-17 09:01:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"91f1532c8a9239b3fba5f1d80b136ea7982a501f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:01:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1f3f139e6588220c09718c9bb5ac47a24430d856","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:10:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/536cf7d9fcfd4188b8e6722a2088f77a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b14cb48c1a142dda00825b4f5cdd099 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91084f2ffc3b47a9be1207caaca142c9 : SUCCESS in 6m 54s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/de8be1e21372497dbc43e769669cd851 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"07fe88af36790591156f50166cc9a693711e5762","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:10:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"813481e5b311a88a2cf2ea2e349be6bd62b12762","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-18 17:26:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"78ab2779555734e3d67287f20693077159bf9b74","revisions":{"abc364046a1672fcaf58eb55e229bead9c4ba1c5":{"kind":"REWORK","_number":1,"created":"2025-04-14 19:04:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/947147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/947147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/947147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/947147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/947147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/947147/1"}}},"commit":{"parents":[{"commit":"775bfa001f7d2a3c1a7c4026b59c776d5f927943","subject":"Merge \"Updating the Octavia and Designate liaisons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/775bfa001f7d2a3c1a7c4026b59c776d5f927943"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-04-14 19:03:36.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-04-14 19:04:07.000000000","tz":0},"subject":"Release Octavia stable/2023.2 bugfixes","message":"Release Octavia stable/2023.2 bugfixes\n\nChange-Id: I0634d4045bfaa12790e04aef745b8612f550af18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abc364046a1672fcaf58eb55e229bead9c4ba1c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abc364046a1672fcaf58eb55e229bead9c4ba1c5"}]},"branch":"refs/heads/master"},"78ab2779555734e3d67287f20693077159bf9b74":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-14 19:06:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/947147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/947147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/947147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/947147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/947147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/947147/2"}}},"commit":{"parents":[{"commit":"775bfa001f7d2a3c1a7c4026b59c776d5f927943","subject":"Merge \"Updating the Octavia and Designate liaisons\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/775bfa001f7d2a3c1a7c4026b59c776d5f927943"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-04-14 19:03:36.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-04-14 19:06:18.000000000","tz":0},"subject":"Release Octavia stable/2023.2 bugfixes","message":"Release Octavia stable/2023.2 bugfixes\n\n$ git log --oneline --no-merges 13.0.0..HEAD\n3142d13ab Cleanup duplicate DB sessions for quotas\n37e6e6395 Fix amphora image builds to use DIB bindep\nc972ff9b8 Fix port abandonment on failover flow revert\n5a23a8eca Fixed VRRP nopreempt option\n2bae2e287 Fix test_prometheus_proxy.test_main\n3497e769a Fix incorrect filtering when using bool attrs\nce0066f03 Handle undefined protocol field in security group rules correctly\n35817025b Update stable/2023.2 to use 2023.2 tests\nf1ea5ed2c Fix verification of certificates signed by a private CA\n6987845c8 Fix missing GARP with UDP listeners on SINGLE LB\nceb55b567 Remove amphora_health record on revert CreateAmphoraInDB\n649f8967b Do not fail on revert PlugVIPAmphora due undefined db_lb\nb4adc98f6 Use cryptography to load PKCS12 certificates\n09ea1e991 Fix loadbalancer stuck in cascade delete\n70ecbcc8b Add check for duplicate members in batch update\n80bbf2e56 Removing tips jobs on stable/2023.2\ndbe59b8b4 When we failed to load pkcs12 cert print warning\nf548fa53a Fix DB deadlock in quotas\n2f170b79a Fix health monitor information retrieval in API response\n1f38d29b2 Fix incorrect masquerade rules in multivip LBs\na9a3c64ee Fix neutron setting overrides\na1d422fe7 Fix fully-populated API with allowed_cidrs\n3142d13ab Cleanup duplicate DB sessions for quotas\n37e6e6395 Fix amphora image builds to use DIB bindep\nc972ff9b8 Fix port abandonment on failover flow revert\n5a23a8eca Fixed VRRP nopreempt option\n2bae2e287 Fix test_prometheus_proxy.test_main\n3497e769a Fix incorrect filtering when using bool attrs\nce0066f03 Handle undefined protocol field in security group rules correctly\n35817025b Update stable/2023.2 to use 2023.2 tests\nf1ea5ed2c Fix verification of certificates signed by a private CA\n6987845c8 Fix missing GARP with UDP listeners on SINGLE LB\nceb55b567 Remove amphora_health record on revert CreateAmphoraInDB\n649f8967b Do not fail on revert PlugVIPAmphora due undefined db_lb\nb4adc98f6 Use cryptography to load PKCS12 certificates\n09ea1e991 Fix loadbalancer stuck in cascade delete\n70ecbcc8b Add check for duplicate members in batch update\n80bbf2e56 Removing tips jobs on stable/2023.2\ndbe59b8b4 When we failed to load pkcs12 cert print warning\nf548fa53a Fix DB deadlock in quotas\n2f170b79a Fix health monitor information retrieval in API response\n1f38d29b2 Fix incorrect masquerade rules in multivip LBs\na9a3c64ee Fix neutron setting overrides\na1d422fe7 Fix fully-populated API with allowed_cidrs\nb509b0795 Handle empty delay on update healthmonitor\n06142805e Fix negative or 0 limit parameter in pagination\n56cd4d3a9 Pin pylint (\u003c\u003d3.0.4)\n33627fbb4 Fix error when deleting LB with broken amp\nc664c865b fix: specify endpoint info. for neutron client\nacf4a3064 fix pep8 use-yield-from rule\nd0725c084 Fix issue with certificates with no subject or CN\n421c7fdce Stable-only: Cap hacking to \u003c 6.1.0\n0f9bc3b8c Fix TLS-HELLO healthmonitors in the amphora-driver\n7c468ca4c Fix health-monitors with ALPN members\n281379c6d Remove publish-openstack-octavia-amphora-image jobs\nbe828d7cf Remove slqalchemy-tips job\n88d7315a6 Add Noop Certificate Manager\n7c67c1023 Fix Amphora Configure API call\nef4b4d500 Retry to set loadbalancer prov status on failures\n562243153 Fix amphorae in ERROR during the failover\n6fdc2079d Reduce duration of failovers with amphora in ERROR\n54eb1624f Fix timeout duration in start_vrrp_service during failovers\na04e0f520 [stable/2023.2] Remove octavia-grenade-skip-level job\n8a5bed4ce Fix race condition in members batch update API call\n4f827c4a7 Fix remaining usage of [neutron] endpoint_type\nf5d56c0fd Update TOX_CONSTRAINTS_FILE for stable/2023.2\n853d0f69e Update .gitreview for stable/2023.2\n\nChange-Id: I0634d4045bfaa12790e04aef745b8612f550af18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78ab2779555734e3d67287f20693077159bf9b74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78ab2779555734e3d67287f20693077159bf9b74"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
