)]}'
{"id":"openstack%2Freleases~948447","triplet_id":"openstack%2Freleases~master~If2e3b246741f5139ecec579c232a5cc03f7a2fb8","project":"openstack/releases","branch":"master","topic":"neutron-2023.2-23.4.1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-04-29 16:09:12.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-04-29 16:09:12.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-04-29 15:50:43.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2025-04-29 16:09:12.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2025-04-29 16:09:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2e3b246741f5139ecec579c232a5cc03f7a2fb8","subject":"Release neutron for stable/2023.2","status":"MERGED","created":"2025-04-29 15:00:37.000000000","updated":"2025-04-29 16:09:12.000000000","submitted":"2025-04-29 16:09:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948447-neutron-2023.2-23.4.1","meta_rev_id":"b35202a5b142cb8e6b74a343b659c3034c318213","_number":948447,"virtual_id_number":948447,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 16:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-04-29 15:29:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-04-29 15:50:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-04-29 15:50:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-04-29 15:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-29 15:01:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-29 15:08:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-04-29 15:20:05.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2025-04-29 15:22:35.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-04-29 15:26:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-04-29 15:29:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3ed2b556cc76e318fd62e34b3fd440ba8a80bd76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-29 15:00:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce71b07d9e405ea80dd8e53fe6485b4eabefe227","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:01:12.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"353938ddaf602a990b1a23238daa37da5d1a41e1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-04-29 15:08:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2eff7207c8b94eb4e8ea9ee00b1b37c64c6d8d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:16:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13bdfebf11e8492285345f1c1467ecb6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae4bda15beb4a96ba393750745e1c1e : SUCCESS in 7m 40s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/16584f26e03d4b12b4b84dd597798e48 : SUCCESS in 8m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8d82c69fab9049bdb339f70953a67a96 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"979a045d8d327264f487f126be997a6f26fa5b64","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-29 15:22:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e34d926e5795004d20a9111b06cd38dfd05d5f1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-29 15:26:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86a00d99c9815f1ea5000deb1f70931192103960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-29 15:29:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"175dd2c5754560cdc417662eaea3d1c1e941cfe9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-04-29 15:29:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e5d13fe8d9ef71a89a95b388b71fb2e4cc72ebb5","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:29:40.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"95a4cf9745c12a7dbee6a5bd5ca955828e7442fe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-29 15:29:46.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d6d0520a69c41110ac8a03e3fa384611390e154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:43:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00eac2e9307746e3bde1419f5baedc4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899bdaa2892f4474987db2f33e0814e1 : SUCCESS in 7m 59s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/aa98508796d34728a60a6bf458b2a0a5 : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f9fcc10f202d445bb52b46578322ecd0 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4db6954ebe4f2367af371f4329260365c41e84d4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-04-29 15:50:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"637d6decd48f2f8e271c45f82fbfab0432d146a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 15:50:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"16264046523d314eeebe9af1a9c0a7e66e5e96f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 16:09:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318ba60ee9fe4d0986c3373743c60a3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0f7b07586c43a981cbea5a8d180eac : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5e7f667d0c4fcf8357bb420ab2caa2 : SUCCESS in 6m 57s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/e87b651beb6640f789ea3a6f5b62714f : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b35202a5b142cb8e6b74a343b659c3034c318213","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 16:09:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14bebec91df640805dbf44f61463e27480ad10d9","revisions":{"bd32798cc63464552e0aa8313a9634d8c8eaacf0":{"kind":"REWORK","_number":1,"created":"2025-04-29 15:00:37.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/47/948447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/948447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/948447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/948447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/948447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/948447/1"}}},"commit":{"parents":[{"commit":"8a3a3ba5af6666e684881e75225da8108edabe86","subject":"Merge \"Final cinder project 2023.2 (Bobcat) releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a3a3ba5af6666e684881e75225da8108edabe86"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-04-29 14:58:02.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-04-29 14:58:02.000000000","tz":-240},"subject":"Release neutron for stable/2023.2","message":"Release neutron for stable/2023.2\n\nThis is the last release of neutron for 2023.2 (Bobcat)\nbefore transitioning to End of Life.\n\n$ git log --oneline --no-merges 23.4.0..HEAD\na4bf5b781a [UT] Delete the OVN AgentCache instance in the cleanup\n647fbbd5ce Create random named resources in ``TestOvnSbSync``\nab3b600a62 [OVN] Isolate test_ovn_client* Functional Tests\nddd09a9584 [OVN][FT] Use MySQL backend for ``TestNBDbMonitor*`` classes\ncb222b442c [OVN] Isolate ``test_ovsdb_monitor.TestNBDbMonitor*`` FTs\nd4b18a263f [OVN] Isolate ``test_maintenance.Test*Maintenance`` FTs\nbbc5ac38d6 Remove the duplicated YOGA database milestone\n2fd1dec9c5 [stable-only] Fix the Neutron milestones list\nb98de1f0cd [OVN] Isolate test_ovn_db_sync.TestOvnNbSync.* FTs\nbe59c0c0f1 [S-RBAC] Fix policies for the SG rules API\n5b69cc5220 Don\u0027t change original target dict by the OwnerCheck policy rule\ne1fa4c18fa Fast exit if \"segments\" plugin is not loaded\n588fdf5e28 [OVN] Do not delete twice the agent from the cache\na4ec5479d7 [OVN] Use the MySQL backend for the ``TestOvnNbSync`` tests (2)\n4c7f9872ca [ovn][trivial] Add \u0027empty_string_filtering\u0027 extension to OVN\n834681659f [QoS] QoS rule check also considers the direction\n\nChange-Id: If2e3b246741f5139ecec579c232a5cc03f7a2fb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd32798cc63464552e0aa8313a9634d8c8eaacf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bd32798cc63464552e0aa8313a9634d8c8eaacf0"}]},"branch":"refs/heads/master"},"14bebec91df640805dbf44f61463e27480ad10d9":{"kind":"REWORK","_number":2,"created":"2025-04-29 15:29:04.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/47/948447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/47/948447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/948447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/948447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/47/948447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/47/948447/2"}}},"commit":{"parents":[{"commit":"8a3a3ba5af6666e684881e75225da8108edabe86","subject":"Merge \"Final cinder project 2023.2 (Bobcat) releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8a3a3ba5af6666e684881e75225da8108edabe86"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-04-29 14:58:02.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-04-29 15:28:36.000000000","tz":-240},"subject":"Release neutron for stable/2023.2","message":"Release neutron for stable/2023.2\n\nThis is the last release of neutron for 2023.2 (Bobcat)\nbefore transitioning to End of Life.\n\n$ git log --oneline --no-merges 23.4.0..HEAD\na4bf5b781a [UT] Delete the OVN AgentCache instance in the cleanup\n647fbbd5ce Create random named resources in ``TestOvnSbSync``\nab3b600a62 [OVN] Isolate test_ovn_client* Functional Tests\nddd09a9584 [OVN][FT] Use MySQL backend for ``TestNBDbMonitor*`` classes\ncb222b442c [OVN] Isolate ``test_ovsdb_monitor.TestNBDbMonitor*`` FTs\nd4b18a263f [OVN] Isolate ``test_maintenance.Test*Maintenance`` FTs\nbbc5ac38d6 Remove the duplicated YOGA database milestone\n2fd1dec9c5 [stable-only] Fix the Neutron milestones list\nb98de1f0cd [OVN] Isolate test_ovn_db_sync.TestOvnNbSync.* FTs\nbe59c0c0f1 [S-RBAC] Fix policies for the SG rules API\n5b69cc5220 Don\u0027t change original target dict by the OwnerCheck policy rule\ne1fa4c18fa Fast exit if \"segments\" plugin is not loaded\n588fdf5e28 [OVN] Do not delete twice the agent from the cache\na4ec5479d7 [OVN] Use the MySQL backend for the ``TestOvnNbSync`` tests (2)\n4c7f9872ca [ovn][trivial] Add \u0027empty_string_filtering\u0027 extension to OVN\n834681659f [QoS] QoS rule check also considers the direction\n\nChange-Id: If2e3b246741f5139ecec579c232a5cc03f7a2fb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14bebec91df640805dbf44f61463e27480ad10d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14bebec91df640805dbf44f61463e27480ad10d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
