)]}'
{"id":"openstack%2Freleases~951637","triplet_id":"openstack%2Freleases~master~Ia92183f65b0a3e08893fbcb93f74ff71f82294e1","project":"openstack/releases","branch":"master","topic":"new-observabilityclient","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-06-04 19:07:07.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2025-06-05 09:06:37.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"5241":{"account":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"last_update":"2025-06-05 12:57:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia92183f65b0a3e08893fbcb93f74ff71f82294e1","subject":"Release new version of python-observabilityclient","status":"MERGED","created":"2025-06-03 09:26:02.000000000","updated":"2025-06-05 12:57:23.000000000","submitted":"2025-06-05 12:57:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951637-new-observabilityclient","meta_rev_id":"84c09b6557edfe6debc1d7f1e1612ab21d3c43fc","_number":951637,"virtual_id_number":951637,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-05 12:57:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2025-06-05 12:47:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-06-04 19:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2025-06-05 09:06:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2025-06-05 12:47:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2025-06-04 15:31:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 09:26:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 11:20:09.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2025-06-03 12:08:28.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2025-06-05 12:47:18.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"51d1d119424aeff10f4e0a2a34c3665c2c4bad09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2025-06-03 09:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86d833d6797727ee61b1985328e8c4601a48867f","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 09:26:43.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"b6ed5f3559704fe4ee4c18702d49bc7e3b6b0045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 09:36:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ef57209af05426996f870eba8efc8bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fdbc86032244f2bbb2c748289f9d4ae : SUCCESS in 5m 25s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ea58d4c311b1443abd1796b8b41c1b2b : SUCCESS in 4m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b837dc656de640909f021d0687433d4a : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a1c3e9bdba5a8cbf5af0a027f6434017e4239d2f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-06-03 12:08:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fc12e5c57617b60d48efdfa6cf912083f17e0e1b","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2025-06-04 12:14:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0052fc385a3559ceaaade63473a4ff683ae6279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2025-06-04 15:31:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ff5e095e2f920bf9547b7bdef5e152cba7980d66","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2025-06-04 15:31:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d070b165abb612bdb8b1640d9166fac93f3da9f1","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 15:31:37.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"f8adc4594558d182b77915d616a597153924d1d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 15:37:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b4322e4f65949cfbb2a5782ac2f48d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea9d4943f9124d76951cf58a715cc7c0 : SUCCESS in 5m 33s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/1a424cbefab742318103a9109e4cc8b3 : SUCCESS in 4m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5dcd426e06794e6f92080ef035ba8380 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"3b3cf6b76f9a43a79638fba6fcec4c9dc1bbf61b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-06-04 19:07:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8621ed7caa3e4a3c5733f6a1dcc4c030a6fe0aa","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2025-06-05 09:06:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1bb3a9aec3304de6f49f4b53bb110d9951542462","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-06-05 12:47:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9e942544f1aeb31684e9ba00b9f7ac3b806cc851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 12:47:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa6b11a82261e1bce40a79c2149902df34aaa925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 12:57:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918a4c02921d4d8abd7df8687bc6fb51\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd03f438952f45348978ced92f23c941 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b16470a308a46e0a062533cf6a040b8 : SUCCESS in 6m 01s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/ebd5e483ded14d55a54bcfa279097e8b : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"84c09b6557edfe6debc1d7f1e1612ab21d3c43fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 12:57:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5548f8416c8cfc550d23beb761cdc2c833017c93","revisions":{"176b075a4f8a56fcb842f7cbf8d1e8fd9b93a608":{"kind":"REWORK","_number":1,"created":"2025-06-03 09:26:02.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/37/951637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/951637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/951637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/951637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/951637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/951637/1"}}},"commit":{"parents":[{"commit":"5d02b055fbd289fa5b42127d1761128fc3eb6efa","subject":"Merge \"Release osprofiler 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d02b055fbd289fa5b42127d1761128fc3eb6efa"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2025-06-03 09:23:02.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2025-06-03 09:25:56.000000000","tz":120},"subject":"Release new version of python-observabilityclient","message":"Release new version of python-observabilityclient\n\nThis release starts to move tenant awareness functionality of Prometheus\nqueries from the client itself towards new project Aetos.\n\nChange-Id: Ia92183f65b0a3e08893fbcb93f74ff71f82294e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/176b075a4f8a56fcb842f7cbf8d1e8fd9b93a608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/176b075a4f8a56fcb842f7cbf8d1e8fd9b93a608"}]},"branch":"refs/heads/master"},"5548f8416c8cfc550d23beb761cdc2c833017c93":{"kind":"REWORK","_number":2,"created":"2025-06-04 15:31:02.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/37/951637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/37/951637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/951637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/951637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/37/951637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/37/951637/2"}}},"commit":{"parents":[{"commit":"5d02b055fbd289fa5b42127d1761128fc3eb6efa","subject":"Merge \"Release osprofiler 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5d02b055fbd289fa5b42127d1761128fc3eb6efa"}]}],"author":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2025-06-04 15:30:51.000000000","tz":120},"committer":{"name":"Martin Mágr","email":"mmagr@redhat.com","date":"2025-06-04 15:30:51.000000000","tz":120},"subject":"Release new version of python-observabilityclient","message":"Release new version of python-observabilityclient\n\nThis release starts to move tenant awareness functionality of Prometheus\nqueries from the client itself towards new project Aetos.\n\nChange-Id: Ia92183f65b0a3e08893fbcb93f74ff71f82294e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5548f8416c8cfc550d23beb761cdc2c833017c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5548f8416c8cfc550d23beb761cdc2c833017c93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
