)]}'
{"id":"openstack%2Freleases~954342","triplet_id":"openstack%2Freleases~master~Ib9a4c2e85a91895983cea52b5f75b0cdba1cd4bc","project":"openstack/releases","branch":"master","topic":"tool-delete-branch","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2025-07-25 12:01:38.000000000","reason":"\u003cGERRIT_ACCOUNT_308\u003e replied on the change","reason_account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-07-10 13:03:30.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-07-25 12:19:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9a4c2e85a91895983cea52b5f75b0cdba1cd4bc","subject":"[tool] Fix delete_*_branch.py","status":"MERGED","created":"2025-07-08 14:02:22.000000000","updated":"2025-07-25 12:19:35.000000000","submitted":"2025-07-25 12:19:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954342-tool-delete-branch","meta_rev_id":"28c666a0f886bc411237bdde7df6b27c132f4bdc","_number":954342,"virtual_id_number":954342,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-25 12:19:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2025-07-25 12:01:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2025-07-10 13:03:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2025-07-25 12:01:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2025-07-10 13:03:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-08 14:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-10 09:11:38.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2025-07-10 09:59:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-07-10 13:03:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-07-10 18:50:14.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"608555436f43b40d229001ab85da115c099b1fb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-08 14:02:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32c129d37b3ae25dc60f3a65780a4ad1f92bc45c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-08 14:16:05.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d558e844c614f548f4b5d32995d1ec4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b8383730aa4631b70b827bb9bd07d0 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8021f7c9cfc1436ca574373656de6441 : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b4f451ac6ef43f99af7bfa849b6f0fe : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89d5ca8a51ce4c3f9202cfa4e3acbb4c : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ecf07905468147ce910fb805929255a1 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15b0ee01a06148ed9ef022ba9fa8bd05 : SUCCESS in 5m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27a96789ace84a579d81725708f2d6c2 : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7ed255f67f402baf77745f3d2fd55d : SUCCESS in 6m 59s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/d5378a0b464c4c788c54872c91962aa0 : SUCCESS in 2m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9edab8c176e747adb558c05a9b64fd69 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2017b5afa26f89f37827dd18c9431639d19caeca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-09 08:27:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e63581a9cbccc313d507a338e80061fb9e32e769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-09 08:42:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038c29f8b91742dda99f8e37d5088c11\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be6d80f5cdbb486cbf79f97cec1bc281 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfc1ebfdab354e8184ff7fcd1d40974a : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05d27695d76f4abbb073a12b5a07438e : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/530a7d1459154f0086b3d98144b33966 : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c793c83d0e4547c5b3e34c5d6dc7c292 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f7c436afbce4c2bbe9665b7c02950f9 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb25f07b375940d2a4aa976ce75a7195 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6034cff4b842459090eb003a625d04 : SUCCESS in 7m 47s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/6b510c71b3334d1795d27017d6dba820 : SUCCESS in 4m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/022bc9dce8e84194b2c5c65229c3c14e : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"189c6e8701f210fde430291fbeec16727116eec4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-07-10 09:11:38.000000000","message":"Patch Set 2: Code-Review+2 PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"26ceb9d0610c54d61964ca6f1dc3ceab719af287","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 09:59:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd75ec52768fd919e52afefbb0594d25289a1b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-10 11:10:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* PTL-Approved+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f23dfdd04a95cadb83e74eb5d1143aa7bc68462e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-07-10 11:11:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43e20d559d630f088de45e2690088fc5ea98bb48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-10 11:35:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78b2caed69643a88f0fb824d71edca4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfba9f45de8429c83fb54ae5287ac88 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aebf34023c1c46f387cd91a78224a666 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15b34863a1ea4ed9bf33460aafea36d7 : SUCCESS in 7m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8518a2896cc94bfabc267afd624e0fbe : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/690112c848d74c3192348e84b83fc1c6 : SUCCESS in 5m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50859c3c58d4406e876b56316c1ced32 : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ca9dd8fc5e8b4b66b4a226723d6c58f0 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f23fc485f9734a18b796a4dc47a6e7c0 : SUCCESS in 6m 43s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/184b143d8f83471cb371b5d76214b33a : SUCCESS in 5m 36s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/34d4e175e36d40abb9d462b88d9b8dd2 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"400765b6e2b46c6713b9e5d06f7211ddee922c57","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-07-10 13:03:30.000000000","message":"Patch Set 3: Code-Review+2 PTL-Approved+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"04a864de818afeb63550743c2c0f05e8f22945fe","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-07-10 18:50:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27b2d698c4bcd11ef3974a6a6579bcbe37713501","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2025-07-25 12:01:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"60353031c038a755907fa0fc6cc94cfc15b8f7d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 12:02:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"eafaeecf30b1ea3f830fcfc4a8dc5b18611dac34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 12:19:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72bc51642cf8487f9c0f9866eef1b8c5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e00ae4b6379b4918b89c8f694ac9a2e0 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc223de885e44838bf9258fd045d12c0 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d68d5c68d6244b1803d33c08152497b : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e22b8ed6284944469d439f8d173b014f : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d92a0b4cccf943c195bfa5231b8750dc : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cea1e9274bf94a0b96e7b150af619ef0 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed3f11f4011041bdb1e2de958d66ff24 : SUCCESS in 6m 05s\n- releases-tox-validate https://zuul.opendev.org/t/openstack/build/9574ab3b71964626bc901e9ef48d90d6 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"28c666a0f886bc411237bdde7df6b27c132f4bdc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 12:19:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2ca41a21832cc9ed0cd9a7aff04f6919d4c0ffe7","revisions":{"97ebeade8cade12461b0f7cacd26150c59c1d44a":{"kind":"REWORK","_number":1,"created":"2025-07-08 14:02:22.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/42/954342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/954342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/954342/1"}}},"commit":{"parents":[{"commit":"23600cbc823de27e098db1b8ffe9925caf91f19b","subject":"Merge \"[Puppet OpenStack] Transition 2023.2 Bobcat to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23600cbc823de27e098db1b8ffe9925caf91f19b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-07-08 13:56:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-07-08 14:02:08.000000000","tz":120},"subject":"[tool] Fix delete_*_branch.py","message":"[tool] Fix delete_*_branch.py\n\nGitea changed the 404 response json as previously it gave back an\n\u0027error\u0027: [\u0027branch does not exist [name: stable/{branch_id}\u0027]\nin a json dict, but now only returns\n\u0027error\u0027: null\n\nChange-Id: Ib9a4c2e85a91895983cea52b5f75b0cdba1cd4bc\nSigned-off-by: Előd Illés \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97ebeade8cade12461b0f7cacd26150c59c1d44a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/97ebeade8cade12461b0f7cacd26150c59c1d44a"}]},"branch":"refs/heads/master"},"3f6a5fd4616fdca9eaa1d152819afacd7d4d7811":{"kind":"REWORK","_number":2,"created":"2025-07-09 08:27:08.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/42/954342/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/954342/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/954342/2"}}},"commit":{"parents":[{"commit":"23600cbc823de27e098db1b8ffe9925caf91f19b","subject":"Merge \"[Puppet OpenStack] Transition 2023.2 Bobcat to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23600cbc823de27e098db1b8ffe9925caf91f19b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-07-08 13:56:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-07-09 08:26:57.000000000","tz":120},"subject":"[tool] Fix delete_*_branch.py","message":"[tool] Fix delete_*_branch.py\n\nGitea changed the 404 response json as previously it gave back an\n\u0027error\u0027: [\u0027branch does not exist [name: stable/{branch_id}\u0027]\nin a json dict, but now only returns\n\u0027error\u0027: null\n\nChange-Id: Ib9a4c2e85a91895983cea52b5f75b0cdba1cd4bc\nSigned-off-by: Előd Illés \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3f6a5fd4616fdca9eaa1d152819afacd7d4d7811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3f6a5fd4616fdca9eaa1d152819afacd7d4d7811"}]},"branch":"refs/heads/master"},"2ca41a21832cc9ed0cd9a7aff04f6919d4c0ffe7":{"kind":"REWORK","_number":3,"created":"2025-07-10 11:10:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/42/954342/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/954342/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/954342/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/954342/3"}}},"commit":{"parents":[{"commit":"23600cbc823de27e098db1b8ffe9925caf91f19b","subject":"Merge \"[Puppet OpenStack] Transition 2023.2 Bobcat to End of Life\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/23600cbc823de27e098db1b8ffe9925caf91f19b"}]}],"author":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-07-08 13:56:10.000000000","tz":120},"committer":{"name":"Előd Illés","email":"elod.illes@est.tech","date":"2025-07-10 11:09:56.000000000","tz":120},"subject":"[tool] Fix delete_*_branch.py","message":"[tool] Fix delete_*_branch.py\n\nGitea changed the 404 response json as previously it gave back an\n\u0027error\u0027: [\u0027branch does not exist [name: stable/{branch_id}\u0027]\nin a json dict, but now only returns\n\u0027error\u0027: null\n\nChange-Id: Ib9a4c2e85a91895983cea52b5f75b0cdba1cd4bc\nSigned-off-by: Előd Illés \u003celod.illes@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ca41a21832cc9ed0cd9a7aff04f6919d4c0ffe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2ca41a21832cc9ed0cd9a7aff04f6919d4c0ffe7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"PTL-Approved","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"PTL-Approved","description":"PTL Approval","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:PTL-Approved\u003e0","fulfilled":true,"status":"PASS","passing_atoms":["label:PTL-Approved\u003e0"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
